)]}'
{"id":"openstack%2Fproject-config~711257","triplet_id":"openstack%2Fproject-config~master~I1b2b9afc9286a23e5eaadc3a67accfa848530d99","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I1b2b9afc9286a23e5eaadc3a67accfa848530d99","subject":"Revert \"Disable limestone\"","status":"MERGED","created":"2020-03-04 15:55:59.000000000","updated":"2020-03-11 20:25:05.000000000","submitted":"2020-03-11 20:25:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711257-1583958305935-660eed34","meta_rev_id":"d64d43c6fceee1372fa0b85f468a500a950b2e57","_number":711257,"virtual_id_number":711257,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-11 20:25:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2020-03-11 20:13:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2020-03-04 15:56:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-03-11 18:16:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2020-03-11 20:13:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-04 15:56:33.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-03-11 18:16:52.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2020-03-11 20:25:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a101d390d0102bb8868919696890dbee868d538","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-03-04 15:55:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0463d0ae9a4adee3f7ff8b505690ec383207c43","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-04 15:56:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ac685466d7dc034b57d582e3c9383b5cfa12b661","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 16:28:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e2bb5d7371c54b7eb5394fe30d5ce55f : SUCCESS in 7m 29s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/5e50a92344c44b019d9457bd9947016e : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":1},{"id":"88426b74a7b6c0ffc253b5e2382b6cf15eeb5f99","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-03-05 15:40:50.000000000","message":"Patch Set 1: Workflow-1\n\nMirror is still offline.","accounts_in_message":[],"_revision_number":1},{"id":"c4185afa65fb0bf0e4700d194e17f7d81d98c796","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2020-03-11 18:16:52.000000000","message":"Patch Set 1: Code-Review+1\n\nthis is resolved now. the node hosting the mirror had thermal issues which our support folks fixed up. mirror seems to be responding normally now at http://mirror.regionone.limestone.openstack.org/","accounts_in_message":[],"_revision_number":1},{"id":"3512d1ad5486eed052d388c8e14de9fd4b0dc4aa","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-03-11 20:13:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cb515203257ff557328d1c1bcf9f2ac3a04012a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 20:14:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c63093a480951cff8691adf349ffb2236e3b3302","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 20:25:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1d4bda72b55a4221989854d5b3d68fb8 : SUCCESS in 7m 46s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/72ef80d89a2d45608bb5cb0385c6979c : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"fb1f9cc73f03c19604cf422f2426682469041af0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 20:25:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d247cc8012716f0127ef9d02fc80a748e7e07ac9","revisions":{"d247cc8012716f0127ef9d02fc80a748e7e07ac9":{"kind":"REWORK","_number":1,"created":"2020-03-04 15:55:59.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/57/711257/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/711257/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/711257/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/711257/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/711257/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/711257/1"}}},"commit":{"parents":[{"commit":"30557a4c4918bcc1b752712bca592a190b389e10","subject":"Disable limestone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/30557a4c4918bcc1b752712bca592a190b389e10"}]}],"author":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2020-03-04 15:55:38.000000000","tz":-480},"committer":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2020-03-04 15:55:48.000000000","tz":-480},"subject":"Revert \"Disable limestone\"","message":"Revert \"Disable limestone\"\n\nThis reverts commit 30557a4c4918bcc1b752712bca592a190b389e10.\n\nChange-Id: I1b2b9afc9286a23e5eaadc3a67accfa848530d99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d247cc8012716f0127ef9d02fc80a748e7e07ac9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d247cc8012716f0127ef9d02fc80a748e7e07ac9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
