)]}'
{"id":"openstack%2Fproject-config~711288","triplet_id":"openstack%2Fproject-config~master~I85edf619c734c48cdd58ac7e9ed9e49a8963dade","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I85edf619c734c48cdd58ac7e9ed9e49a8963dade","subject":"Move lodgeit to opendev tenant","status":"MERGED","created":"2020-03-04 17:16:57.000000000","updated":"2020-03-04 18:02:18.000000000","submitted":"2020-03-04 18:02:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711288-1583344938038-72e87165","meta_rev_id":"16ffb928205d91c46ab0d3b970e770a05407bb38","_number":711288,"virtual_id_number":711288,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-04 18:02:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2020-03-04 17:49:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2020-03-04 17:17:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2020-03-04 17:49:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-04 17:17:39.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-03-04 17:49:13.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2020-03-04 18:02:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"18816fc24d7f41121b92bd916158038099c8c8c9","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-04 17:16:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68dc3e3e38bda2d07e8be3bc3207d25f9e3cf31d","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-03-04 17:17:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1895de8f466f270efe70b5e861355e6456d4aea8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 17:31:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/30fd5a4992bd4f2f965aed7a77aa9997 : SUCCESS in 8m 42s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4b220599f5fe4df5868551bbf6cf31e2 : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":1},{"id":"974527ac46ece4f58c6042464ad15ffe4dd43f5e","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-03-04 17:49:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"90a369224689ce9e30f960f94c83b09a686b2b22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 17:49:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"71f366f8ddf9ee6814da3dc3169459c6bf3d5618","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 18:02:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e575097885b944ce8384720e780df1fe : SUCCESS in 7m 42s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/76eaa957fb53496f85453c4ac28a5cfd : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"16ffb928205d91c46ab0d3b970e770a05407bb38","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 18:02:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3897c6f2b73a54b8dfd52ab5411159ca62c578f4","revisions":{"3897c6f2b73a54b8dfd52ab5411159ca62c578f4":{"kind":"REWORK","_number":1,"created":"2020-03-04 17:16:57.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/88/711288/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/88/711288/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/711288/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/711288/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/88/711288/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/88/711288/1"}}},"commit":{"parents":[{"commit":"dd2a9a8a02a611fcd000d77bd0d588a8a1075951","subject":"Merge \"Cleanup #opendev gerritbot\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dd2a9a8a02a611fcd000d77bd0d588a8a1075951"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-04 17:16:20.000000000","tz":-360},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-04 17:16:20.000000000","tz":-360},"subject":"Move lodgeit to opendev tenant","message":"Move lodgeit to opendev tenant\n\nWe\u0027ve got someone working on updating it to python3, let\u0027s go ahead\nand move it before it grows more complex.\n\nAlso, remove central job config, let\u0027s put these in tree.\n\nChange-Id: I85edf619c734c48cdd58ac7e9ed9e49a8963dade\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3897c6f2b73a54b8dfd52ab5411159ca62c578f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3897c6f2b73a54b8dfd52ab5411159ca62c578f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
