)]}'
{"id":"openstack%2Fproject-config~711419","triplet_id":"openstack%2Fproject-config~master~I9fc4a14ae21d3b069db1fed12136b6cc787ae599","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I9fc4a14ae21d3b069db1fed12136b6cc787ae599","subject":"zuul.d/projects.yaml: remove x/networking-6wind entry","status":"MERGED","created":"2020-03-05 10:01:44.000000000","updated":"2020-03-05 18:49:13.000000000","submitted":"2020-03-05 18:49:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711419-1583434153019-a42b5d9c","meta_rev_id":"3888ec64d25a42c517641bbd7adfcdc0d11bafec","_number":711419,"virtual_id_number":711419,"owner":{"_account_id":28375,"name":"Valentina Krasnobaeva","email":"valentina.krasnobaeva@6wind.com","username":"valentine-k"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-03-05 14:44:00.000000000","_account_id":28375,"name":"Valentina Krasnobaeva","email":"valentina.krasnobaeva@6wind.com","username":"valentine-k"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-05 18:49:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13869,"name":"Maxime Leroy","email":"maxime.leroy@6wind.com","username":"mleroy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-03-05 18:38:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-03-05 14:45:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28375,"name":"Valentina Krasnobaeva","email":"valentina.krasnobaeva@6wind.com","username":"valentine-k"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-05 15:44:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13869,"name":"Maxime Leroy","email":"maxime.leroy@6wind.com","username":"mleroy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2020-03-05 18:38:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28375,"name":"Valentina Krasnobaeva","email":"valentina.krasnobaeva@6wind.com","username":"valentine-k"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13869,"name":"Maxime Leroy","email":"maxime.leroy@6wind.com","username":"mleroy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":13869,"name":"Maxime Leroy","email":"maxime.leroy@6wind.com","username":"mleroy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28375,"name":"Valentina Krasnobaeva","email":"valentina.krasnobaeva@6wind.com","username":"valentine-k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-05 14:45:14.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-03-05 15:44:39.000000000","updated_by":{"_account_id":13869,"name":"Maxime Leroy","email":"maxime.leroy@6wind.com","username":"mleroy"},"reviewer":{"_account_id":13869,"name":"Maxime Leroy","email":"maxime.leroy@6wind.com","username":"mleroy"},"state":"REVIEWER"},{"updated":"2020-03-05 18:38:46.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-03-05 18:49:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d63e7f20a0d99ed97940e1a67eee3435f8d2166a","author":{"_account_id":28375,"name":"Valentina Krasnobaeva","email":"valentina.krasnobaeva@6wind.com","username":"valentine-k"},"date":"2020-03-05 10:01:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c70161208a814eca309e48d21fe7c82b472239d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 10:02:06.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  while scanning for the next token\nfound character \u0027\\t\u0027 that cannot start any token\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 5658, column 1","accounts_in_message":[],"_revision_number":1},{"id":"49241c7d1301cfc4804dfbdac8991fb4c17739c9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-05 10:07:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e9e378585c9161b1f5ddccba8ac5a54988e43f7d","author":{"_account_id":28375,"name":"Valentina Krasnobaeva","email":"valentina.krasnobaeva@6wind.com","username":"valentine-k"},"date":"2020-03-05 14:05:57.000000000","message":"Abandoned\n\nAbandon, as .zuul.yaml with project\u0027s jobs settings was add in the networking-6wind project tree","accounts_in_message":[],"_revision_number":1},{"id":"e06decb2f645f993646f93ffafdd83ef5e5ac36e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-05 14:21:34.000000000","message":"Patch Set 1:\n\nYou still need to remove the python jobs here...\n\nI propose to restore and change as suggested","accounts_in_message":[],"_revision_number":1},{"id":"e3917e10f1deea7857dc52f36f1dad785a24ea1e","author":{"_account_id":28375,"name":"Valentina Krasnobaeva","email":"valentina.krasnobaeva@6wind.com","username":"valentine-k"},"date":"2020-03-05 14:26:18.000000000","message":"Restored\n\nwe need to remove x/networking-6wind entry from \t\t\tzuul.d/projects.yaml","accounts_in_message":[],"_revision_number":1},{"id":"5ad6e143a93d01a8faba52f706ad10b4e147f3e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 14:26:37.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  while scanning for the next token\nfound character \u0027\\t\u0027 that cannot start any token\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 5658, column 1","accounts_in_message":[],"_revision_number":1},{"id":"68c4bcccebab816fa56b27c87da7d0a7c3eec8af","author":{"_account_id":28375,"name":"Valentina Krasnobaeva","email":"valentina.krasnobaeva@6wind.com","username":"valentine-k"},"date":"2020-03-05 14:35:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6b05105c65d362471d90bc60ffd22e0c25b63525","author":{"_account_id":28375,"name":"Valentina Krasnobaeva","email":"valentina.krasnobaeva@6wind.com","username":"valentine-k"},"date":"2020-03-05 14:35:51.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d138a09b20db59b50d27aadb2469b48507097538","author":{"_account_id":28375,"name":"Valentina Krasnobaeva","email":"valentina.krasnobaeva@6wind.com","username":"valentine-k"},"date":"2020-03-05 14:44:00.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f2ef006fb332b179d3307db85877f14f540655c8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-05 14:45:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"264e2a056274c5f73d4bb7c33eee4af06fafe357","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 15:07:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7b3a0f2637a943ac931f52c51f0552df : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":3},{"id":"c8bce290b48385eda429cc00d953a7bffc0db4f1","author":{"_account_id":13869,"name":"Maxime Leroy","email":"maxime.leroy@6wind.com","username":"mleroy"},"date":"2020-03-05 15:44:39.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0316c0f933591b845f5cafce283fe16c011cc88e","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-03-05 18:38:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f2fd314779b58068e76c27706f5055f6851a19f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 18:39:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7afae7357d1eed4af13df3ece2ea22695198f4ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 18:49:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9e8261ac046d4280bfd409828ea8497f : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":3},{"id":"3888ec64d25a42c517641bbd7adfcdc0d11bafec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 18:49:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fe261109f68c74f60dca9b455088817e928e6a56","revisions":{"21fceb4d8f078a4e9a6060da3cb1dc961f856e6c":{"kind":"REWORK","_number":1,"created":"2020-03-05 10:01:44.000000000","uploader":{"_account_id":28375,"name":"Valentina Krasnobaeva","email":"valentina.krasnobaeva@6wind.com","username":"valentine-k"},"ref":"refs/changes/19/711419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/711419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/711419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/711419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/711419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/711419/1"}}},"commit":{"parents":[{"commit":"45573f32850b5dcf4518a4738639dbee1294bc54","subject":"Merge \"Remove base-ozj job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/45573f32850b5dcf4518a4738639dbee1294bc54"}]}],"author":{"name":"Valentina Krasnobaeva","email":"valentina.krasnobaeva@6wind.com","date":"2020-03-04 17:07:04.000000000","tz":60},"committer":{"name":"Valentina Krasnobaeva","email":"valentina.krasnobaeva@6wind.com","date":"2020-03-05 10:01:42.000000000","tz":60},"subject":"Update jobs for  networking-6wind","message":"Update jobs for  networking-6wind\n\nnetworking-6wind no longer supports Python2.7 due to changes in requirements and rpm-deps list\nDue to the following commit in x/networking-6wind:  changes in project requirements list and rpm-deps\n\nChange-Id: I9fc4a14ae21d3b069db1fed12136b6cc787ae599\nSigned-off-by: Valentina Krasnobaeva \u003cvalentina.krasnobaeva@6wind.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/21fceb4d8f078a4e9a6060da3cb1dc961f856e6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/21fceb4d8f078a4e9a6060da3cb1dc961f856e6c"}]},"branch":"refs/heads/master"},"d6634770254475e37c3bc07a3710363696783353":{"kind":"REWORK","_number":2,"created":"2020-03-05 14:35:17.000000000","uploader":{"_account_id":28375,"name":"Valentina Krasnobaeva","email":"valentina.krasnobaeva@6wind.com","username":"valentine-k"},"ref":"refs/changes/19/711419/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/711419/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/711419/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/711419/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/711419/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/711419/2"}}},"commit":{"parents":[{"commit":"45573f32850b5dcf4518a4738639dbee1294bc54","subject":"Merge \"Remove base-ozj job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/45573f32850b5dcf4518a4738639dbee1294bc54"}]}],"author":{"name":"Valentina Krasnobaeva","email":"valentina.krasnobaeva@6wind.com","date":"2020-03-04 17:07:04.000000000","tz":60},"committer":{"name":"Valentina Krasnobaeva","email":"valentina.krasnobaeva@6wind.com","date":"2020-03-05 14:29:01.000000000","tz":60},"subject":"zuul.d/projects.yaml: remove x/networking-6wind entry","message":"zuul.d/projects.yaml: remove x/networking-6wind entry\n\nThe following commit in x/networking-6wind 3b8f98af1c6d\n(\"Fix dup dependency in requirements.txt, align unit tests\")\nadds .zuul.yaml with updated opendev CI Zuul jobs list. So no longer need to\nmaintain x/networking-6wind entry in the global zuul.d/projects.yaml.\n\nChange-Id: I9fc4a14ae21d3b069db1fed12136b6cc787ae599\nSigned-off-by: Valentina Krasnobaeva \u003cvalentina.krasnobaeva@6wind.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d6634770254475e37c3bc07a3710363696783353"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d6634770254475e37c3bc07a3710363696783353"}]},"branch":"refs/heads/master"},"fe261109f68c74f60dca9b455088817e928e6a56":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-03-05 14:35:51.000000000","uploader":{"_account_id":28375,"name":"Valentina Krasnobaeva","email":"valentina.krasnobaeva@6wind.com","username":"valentine-k"},"ref":"refs/changes/19/711419/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/711419/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/711419/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/711419/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/711419/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/711419/3"}}},"commit":{"parents":[{"commit":"45573f32850b5dcf4518a4738639dbee1294bc54","subject":"Merge \"Remove base-ozj job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/45573f32850b5dcf4518a4738639dbee1294bc54"}]}],"author":{"name":"Valentina Krasnobaeva","email":"valentina.krasnobaeva@6wind.com","date":"2020-03-04 17:07:04.000000000","tz":60},"committer":{"name":"Valentina Krasnobaeva","email":"valentina.krasnobaeva@6wind.com","date":"2020-03-05 14:35:35.000000000","tz":60},"subject":"zuul.d/projects.yaml: remove x/networking-6wind entry","message":"zuul.d/projects.yaml: remove x/networking-6wind entry\n\nThe following commit in x/networking-6wind 3b8f98af1c6d\n(\"Fix dup dependency in requirements.txt, align unit tests\")\nadds .zuul.yaml with updated opendev CI Zuul jobs list. So no longer need to\nmaintain x/networking-6wind entry in the global zuul.d/projects.yaml.\n\nChange-Id: I9fc4a14ae21d3b069db1fed12136b6cc787ae599\nSigned-off-by: Valentina Krasnobaeva \u003cvalentina.krasnobaeva@6wind.com\u003e\nAcked-by: Maxime Leroy \u003cmaxime.leroy@6wind.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fe261109f68c74f60dca9b455088817e928e6a56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fe261109f68c74f60dca9b455088817e928e6a56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
