)]}'
{"id":"openstack%2Fproject-config~711474","triplet_id":"openstack%2Fproject-config~master~I2239a70691833354262b17951db6133a8e258998","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I2239a70691833354262b17951db6133a8e258998","subject":"Run openstacksdk functional jobs on ansible 2.8 and 2.9","status":"ABANDONED","created":"2020-03-05 14:58:38.000000000","updated":"2023-06-01 09:18:18.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"836b8e6dec5e91791869b78b11289a82979bda76","_number":711474,"virtual_id_number":711474,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-03-19 14:20:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-03-19 14:00:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-03-19 14:00:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2020-03-19 14:00:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-05 15:37:18.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2020-03-05 18:30:05.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-03-06 08:40:29.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-03-19 13:31:53.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2020-03-19 14:20:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-12 18:39:50.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2022-11-03 21:27:12.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REMOVED"}],"messages":[{"id":"0727e45e38b95828a93533fcc7f8b5bf404966cf","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-05 14:58:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8c089c517eadc3db1b3666953a84bb646bd6d33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 14:59:10.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Job shade-ansible-stable-2.9-functional-devstack not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      name: ansible/ansible\n      default-branch: devel\n      third-party-check:\n        jobs:\n          - openstacksdk-ansible-devel-functional-devstack:\n              branches: devel\n              files:\n                - ^lib/ansible/modules/cloud/openstack/.*\n                - ^contrib/inventory/openstack_inventory.py\n                - ^lib/ansible/plugins/inventory/openstack.py\n                - ^lib/ansible/module_utils/openstack.py\n                - ^lib/ansible/utils/module_docs_fragments/openstack.py\n          - openstacksdk-ansible-stable-2.8-functional-devstack:\n              branches: stable-2.8\n              files:\n                - ^lib/ansible/modules/cloud/openstack/.*\n                - ^contrib/inventory/openstack_inventory.py\n                - ^lib/ansible/plugins/inventory/openstack.py\n                - ^lib/ansible/module_utils/openstack.py\n                - ^lib/ansible/utils/module_docs_fragments/openstack.py\n          - shade-ansible-stable-2.9-functional-devstack:\n              branches: stable-2.9\n              files:\n                - ^lib/ansible/modules/cloud/openstack/.*\n                - ^contrib/inventory/openstack.py\n                - ^lib/ansible/plugins/inventory/openstack.py\n                - ^lib/ansible/module_utils/openstack.py\n                - ^lib/ansible/utils/module_docs_fragments/openstack.py\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 27, column 3","accounts_in_message":[],"_revision_number":1},{"id":"34e39c77fa5c8ec0fa4fed95f14587250bd7f2f2","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-03-05 15:37:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6a3cbdad0c9005b9953753caf92f6058c2c40ce4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 15:45:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/173a25393836473089a9b2553fa3de88 : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":2},{"id":"92ee17e36e3356dc9866f3fd959c8b26cd0a358c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-05 18:30:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5322eade3feb4508a819d46123a47cbb82452b99","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-03-06 08:08:05.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9953268f3289aeb86f865ca31380a82b502d6133","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-03-06 08:40:29.000000000","message":"Patch Set 2: -Code-Review\n\nActually these jobs don\u0027t seem to exist/work anymore, waiting for resolution in the dependency.","accounts_in_message":[],"_revision_number":2},{"id":"a5989c900f20639212310a23200c7d5da2d6cb28","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-06 22:45:23.000000000","message":"Patch Set 2:\n\nfrickler - they should be cleaned up in the dependency - but please check me on that - cause I messed it up before","accounts_in_message":[],"_revision_number":2},{"id":"c8a736115211bc2af729d3714fb49129283c9e4a","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2020-03-19 13:31:53.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ec67693fd7aa84ed1b041e2c7aa5dea96a72771e","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-19 13:58:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cdf4a76ba0a5d0c2a5c66b2af32bea24e7f470f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 13:59:29.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  extra keys not allowed @ data[\u0027third-party-check\u0027][\u0027jobs\u0027][0][\u0027openstacksdk-ansible-devel-functional-devstack\u0027][\u0027required_projects\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      name: ansible/ansible\n      default-branch: devel\n      third-party-check:\n        jobs:\n          - openstacksdk-ansible-devel-functional-devstack:\n              branches: devel\n              files:\n                - ^lib/ansible/modules/cloud/openstack/.*\n                - ^contrib/inventory/openstack_inventory.py\n                - ^lib/ansible/plugins/inventory/openstack.py\n                - ^lib/ansible/module_utils/openstack.py\n                - ^lib/ansible/utils/module_docs_fragments/openstack.py\n              required_projects:\n                - name: openstack/openstacksdk\n                  override-checkout: master\n                - name: openstack/devstack\n                  override-checkout: master\n          - openstacksdk-ansible-stable-2.8-functional-devstack:\n              branches: stable-2.8\n              files:\n                - ^lib/ansible/modules/cloud/openstack/.*\n                - ^contrib/inventory/openstack_inventory.py\n                - ^lib/ansible/plugins/inventory/openstack.py\n                - ^lib/ansible/module_utils/openstack.py\n                - ^lib/ansible/utils/module_docs_fragments/openstack.py\n              required_projects:\n                - name: openstack/openstacksdk\n                  override-checkout: master\n                - name: openstack/devstack\n                  override-checkout: master\n          - openstacksdk-ansible-stable-2.9-functional-devstack:\n              branches: stable-2.9\n              files:\n                - ^lib/ansible/modules/cloud/openstack/.*\n                - ^contrib/inventory/openstack.py\n                - ^lib/ansible/plugins/inventory/openstack.py\n                - ^lib/ansible/module_utils/openstack.py\n                - ^lib/ansible/utils/module_docs_fragments/openstack.py\n              required_projects:\n                - name: openstack/openstacksdk\n                  override-checkout: master\n                - name: openstack/devstack\n                  override-checkout: master\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 27, column 3","accounts_in_message":[],"_revision_number":3},{"id":"405d57db82b3dc5658555215c3ee8538ee0dad22","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-19 14:00:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a3b98a21c15aeb34febf76e7c493c057fe343e86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 14:01:51.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"89d0fdb841db20f5e322003a9314f0e62fab8566","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-03-19 14:04:07.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"5b9a106c250a3d581960ed71ef2ee030be08165b","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2020-03-19 14:09:01.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c1df436f716775788ba1fd3e2312a1b25112c675","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 14:20:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/26d050dc869c48a28617330f8d158bf3 : SUCCESS in 8m 17s","accounts_in_message":[],"_revision_number":4},{"id":"6022087ac851691d73d3236412864cf28d66b2c6","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-03-30 12:26:21.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"98121cc3ca58ec940730abf4584b03c08a64d288","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-07-30 20:05:31.000000000","message":"Removed reviewer Mohammed Naser with the following votes:\n\n* Code-Review+2 by Mohammed Naser \u003cmnaser@vexxhost.com\u003e\n* Workflow+1 by Mohammed Naser \u003cmnaser@vexxhost.com\u003e\n","accounts_in_message":[],"_revision_number":4},{"id":"1e050fbc4081bcc947c4c42f5a13102a41df1702","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-09-03 14:22:26.000000000","message":"Removed Code-Review+1 by Sorin Sbarnea (zbr) \u003cssbarnea@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":4},{"id":"7a95d1cfed569e7fe1fb49494f9653fcd1f967f1","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-09-03 14:22:39.000000000","message":"Patch Set 4: Code-Review-1\n\nThis can be abandoned.","accounts_in_message":[],"_revision_number":4},{"id":"bd3a9edd8b344f4af7425b81072e852a042a7116","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-09-21 12:26:18.000000000","message":"Removed reviewer zbr with the following votes:\n\n* Code-Review-1 by zbr \u003cssbarnea@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":4},{"id":"43266b3fd31959bea85db6fc8634ff8443c0a28e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-01-12 18:39:50.000000000","message":"Patch Set 4:\n\nWhy should this be abandoned? Was it already done in a different change or something?","accounts_in_message":[],"_revision_number":4},{"id":"febb036e5599fb54886de7c508d2854a00e1009c","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2022-11-03 21:27:12.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_10969\u003e with the following votes:\n\n* Code-Review+1 by \u003cGERRIT_ACCOUNT_10969\u003e\n","accounts_in_message":[{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"}],"_revision_number":4},{"id":"836b8e6dec5e91791869b78b11289a82979bda76","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-06-01 09:18:18.000000000","message":"Abandoned\n\nmerge conflict, no response in 2 years, feel free to reopen and update if you think that this is still needed","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bced1f9c0fe181c8974e0fce29ef4dbac3ae7727","revisions":{"ad5f937acf29d5412a29212cc0ff324cb25fc282":{"kind":"REWORK","_number":1,"created":"2020-03-05 14:58:38.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/74/711474/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/74/711474/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/711474/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/711474/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/711474/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/74/711474/1"}}},"commit":{"parents":[{"commit":"3897c6f2b73a54b8dfd52ab5411159ca62c578f4","subject":"Move lodgeit to opendev tenant","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3897c6f2b73a54b8dfd52ab5411159ca62c578f4"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-05 14:58:01.000000000","tz":-360},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-05 14:58:01.000000000","tz":-360},"subject":"Run openstacksdk functional jobs on ansible 2.8 and 2.9","message":"Run openstacksdk functional jobs on ansible 2.8 and 2.9\n\nWe\u0027re only running on 2.5 and 2.6 which are EOL. Run on 2.8 and\n2.9 which are relevant.\n\nDepends-On: https://review.opendev.org/711471\nChange-Id: I2239a70691833354262b17951db6133a8e258998\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ad5f937acf29d5412a29212cc0ff324cb25fc282"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ad5f937acf29d5412a29212cc0ff324cb25fc282"}]},"branch":"refs/heads/master"},"94d9029b4d7c91c7e783d7234a197ffa9243f604":{"kind":"REWORK","_number":2,"created":"2020-03-05 15:37:18.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/74/711474/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/74/711474/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/711474/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/711474/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/711474/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/74/711474/2"}}},"commit":{"parents":[{"commit":"3897c6f2b73a54b8dfd52ab5411159ca62c578f4","subject":"Move lodgeit to opendev tenant","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3897c6f2b73a54b8dfd52ab5411159ca62c578f4"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-05 14:58:01.000000000","tz":-360},"committer":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2020-03-05 15:37:08.000000000","tz":-480},"subject":"Run openstacksdk functional jobs on ansible 2.8 and 2.9","message":"Run openstacksdk functional jobs on ansible 2.8 and 2.9\n\nWe\u0027re only running on 2.5 and 2.6 which are EOL. Run on 2.8 and\n2.9 which are relevant.\n\nDepends-On: https://review.opendev.org/711471\nChange-Id: I2239a70691833354262b17951db6133a8e258998\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/94d9029b4d7c91c7e783d7234a197ffa9243f604"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/94d9029b4d7c91c7e783d7234a197ffa9243f604"}]},"branch":"refs/heads/master"},"89b70efb4cc23b1c7ad269358187a27ab40f5dd7":{"kind":"REWORK","_number":3,"created":"2020-03-19 13:58:42.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/74/711474/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/74/711474/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/711474/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/711474/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/711474/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/74/711474/3"}}},"commit":{"parents":[{"commit":"3897c6f2b73a54b8dfd52ab5411159ca62c578f4","subject":"Move lodgeit to opendev tenant","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3897c6f2b73a54b8dfd52ab5411159ca62c578f4"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-05 14:58:01.000000000","tz":-360},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-19 13:58:33.000000000","tz":-300},"subject":"Run openstacksdk functional jobs on ansible 2.8 and 2.9","message":"Run openstacksdk functional jobs on ansible 2.8 and 2.9\n\nWe\u0027re only running on 2.5 and 2.6 which are EOL. Run on 2.8 and\n2.9 which are relevant.\n\nDepends-On: https://review.opendev.org/711471\nChange-Id: I2239a70691833354262b17951db6133a8e258998\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/89b70efb4cc23b1c7ad269358187a27ab40f5dd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/89b70efb4cc23b1c7ad269358187a27ab40f5dd7"}]},"branch":"refs/heads/master"},"bced1f9c0fe181c8974e0fce29ef4dbac3ae7727":{"kind":"REWORK","_number":4,"created":"2020-03-19 14:00:45.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/74/711474/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/74/711474/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/711474/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/711474/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/711474/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/74/711474/4"}}},"commit":{"parents":[{"commit":"3897c6f2b73a54b8dfd52ab5411159ca62c578f4","subject":"Move lodgeit to opendev tenant","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3897c6f2b73a54b8dfd52ab5411159ca62c578f4"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-05 14:58:01.000000000","tz":-360},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-19 14:00:42.000000000","tz":-300},"subject":"Run openstacksdk functional jobs on ansible 2.8 and 2.9","message":"Run openstacksdk functional jobs on ansible 2.8 and 2.9\n\nWe\u0027re only running on 2.5 and 2.6 which are EOL. Run on 2.8 and\n2.9 which are relevant.\n\nDepends-On: https://review.opendev.org/711471\nChange-Id: I2239a70691833354262b17951db6133a8e258998\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bced1f9c0fe181c8974e0fce29ef4dbac3ae7727"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bced1f9c0fe181c8974e0fce29ef4dbac3ae7727"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
