)]}'
{"id":"openstack%2Fproject-config~711878","triplet_id":"openstack%2Fproject-config~master~I176dc57afcfd627ad470ed8c6cf7fbb7299cbb35","project":"openstack/project-config","branch":"master","topic":"new-project","hashtags":[],"change_id":"I176dc57afcfd627ad470ed8c6cf7fbb7299cbb35","subject":"New repo: devstack-plugin-open-cas","status":"MERGED","created":"2020-03-09 08:55:26.000000000","updated":"2020-03-13 15:51:04.000000000","submitted":"2020-03-13 15:51:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711878-1584114664654-4545ba28","meta_rev_id":"55311a855c7fa51d924f15c237a7170e1e0638db","_number":711878,"virtual_id_number":711878,"owner":{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-03-09 12:02:58.000000000","_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-13 15:51:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2020-03-11 14:29:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-11 17:45:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2020-03-13 15:35:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-03-12 06:38:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},{"value":1,"date":"2020-03-10 16:33:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-03-13 15:35:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-11 17:40:19.000000000","_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-10 16:33:08.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-03-11 14:29:59.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-11 17:40:19.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"REVIEWER"},{"updated":"2020-03-11 17:45:32.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-03-12 06:38:20.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-03-13 15:35:36.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-03-13 15:51:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"063f1a29f4a4a811783f9892f928175b93548c18","author":{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},"date":"2020-03-09 08:55:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6db0a913b9eecf3bcbb9d44b67c5a23c00f77356","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 09:02:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/87434050521845f690660eab6cd91232 : SUCCESS in 6m 44s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/974218d8b5a7489394dc87d3ad3374ca : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"5e9602260e3ae6db0690cb66bc691265e79e242a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-09 09:03:54.000000000","message":"Patch Set 1: Code-Review-1\n\nThis needs a governance change as well, link to it via Needed-By as explained in https://docs.openstack.org/infra/manual/creators.html#adding-the-project-to-the-ci-system\n\nAlso, add IRC notifications.","accounts_in_message":[],"_revision_number":1},{"id":"a5012fcd15e85bd81ecab1e94532e0fd04b10614","author":{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},"date":"2020-03-09 11:17:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9754bd80593782a64b44d65df4bf0ec2206e38a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 11:28:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/55da0567c4fd4009a5dcfb4df966d916 : SUCCESS in 8m 20s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/61cfd5869d7e434bb8d5eafd1352a2b9 : SUCCESS in 3m 39s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/65d6082af83642f4a85d49b1562c7c84 : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":2},{"id":"ff0c1996c9994d2371f080aa257b428f316f9711","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-09 11:43:53.000000000","message":"Patch Set 2:\n\nNote that you will start with an empty repository. Your very first\npatch needs to setup the repository and add all infrastructure so that\nthe jobs pass (this can be done with a cookiecutter project).\n\nAlso, you can import an existing repository including it\u0027s history if\nyou have one. Once the repository is created, importing is not\npossible anymore, changes will have to go through the normal review\nprocess.\n\nIf you have a lot of history to import, please use the upstream field\ninstead of creating a repository and then pushing the patches one at a\ntime. Pushing a large number of related patches all at one time causes\nthe CI infrastructure to slow down, which impacts work on all of the\nother projects using it. After repo creation the infra team will not\npush any content to it; this is your only chance to import directly.\n\nPlease confirm that starting with an empty repo is fine.","accounts_in_message":[],"_revision_number":2},{"id":"391a12fe872baf7edaa2ff053b933fe5218b0153","author":{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},"date":"2020-03-09 12:02:58.000000000","message":"Patch Set 2:\n\nYes, I confirm it is OK to start from an empty repo. I will try to initiate using cookiecutter.\n\nOne question is: how does zuul know this repo, and invoke this plugin? Thanks","accounts_in_message":[],"_revision_number":2},{"id":"a73ae8e59f1aefae12cad883e17b08cf1439c184","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-09 12:05:17.000000000","message":"Patch Set 2:\n\n\u003e One question is: how does zuul know this repo, and invoke this\n \u003e plugin? Thanks\n\nhttps://docs.openstack.org/infra/manual/creators.html explains this.\n\nYou need to add jobs and create/update devstack jobs to use this repo.\n\nChange LGTM.\n\nThis change needs a +1 from the adopting projects\u0027s PTL, I will\nwait for the +1 of the QA PTL before I +2.","accounts_in_message":[],"_revision_number":2},{"id":"9c39ebde630d551190663cb441d4cf18eab5d34d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-10 16:33:08.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks. LGTM.","accounts_in_message":[],"_revision_number":2},{"id":"098a6a1dcdd63eecd389392f9de8b50c40d0925a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-10 18:28:13.000000000","message":"Patch Set 2:\n\nLGTM, Brian, could you review as well, please?","accounts_in_message":[],"_revision_number":2},{"id":"46b24368d491e9474967ee2ae933ecdf07ff430c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-11 14:29:59.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b8bc5b34db88240e7a72b52fbeda99b47200a237","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-03-11 17:45:32.000000000","message":"Patch Set 2: Code-Review+1\n\nSorry for the delay.  LGTM!","accounts_in_message":[],"_revision_number":2},{"id":"9d3c75788ca59d0553c84ab342304361cecab086","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-12 06:38:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b2f5449edcf844a917215ae3036489fbe67a0042","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-03-13 15:35:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1643ffc80feac30e9331d8f638caa1fb4bbb47b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 15:35:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"11cd98961842ad8edac60dd891a3427c1f83e170","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 15:51:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2bd35833a99232e0aaa5ff9b832828cea58eea07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 15:51:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4dd06428acb7461dad1c4cce220db77f : SUCCESS in 7m 06s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/9410ccc84c4b489986bc063cba78d068 : SUCCESS in 3m 47s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/07db7c804fdb4eedb2eb5b7533a7b6f6 : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a26717c6343386e6547621711e2d1e0c7f746e43","revisions":{"e2120ac47c7a1ebde19d02f2654c23338929ff91":{"kind":"REWORK","_number":1,"created":"2020-03-09 08:55:26.000000000","uploader":{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},"ref":"refs/changes/78/711878/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/711878/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/711878/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/711878/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/711878/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/711878/1"}}},"commit":{"parents":[{"commit":"3897c6f2b73a54b8dfd52ab5411159ca62c578f4","subject":"Move lodgeit to opendev tenant","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3897c6f2b73a54b8dfd52ab5411159ca62c578f4"}]}],"author":{"name":"Liang Fang","email":"liang.a.fang@intel.com","date":"2020-03-09 07:09:42.000000000","tz":480},"committer":{"name":"Liang Fang","email":"liang.a.fang@intel.com","date":"2020-03-09 08:36:39.000000000","tz":480},"subject":"New repo: devstack-plugin-open-cas","message":"New repo: devstack-plugin-open-cas\n\nThis repo is to add plugin to configure open-cas which is needed in\nvolume local cache feature.\n\nopen-cas: https://open-cas.github.io/\nNova Spec: https://review.opendev.org/#/c/689070/\nCinder Spec: https://review.opendev.org/#/c/684556/\n\nChange-Id: I176dc57afcfd627ad470ed8c6cf7fbb7299cbb35\nSigned-off-by: Liang Fang \u003cliang.a.fang@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e2120ac47c7a1ebde19d02f2654c23338929ff91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e2120ac47c7a1ebde19d02f2654c23338929ff91"}]},"branch":"refs/heads/master"},"a26717c6343386e6547621711e2d1e0c7f746e43":{"kind":"REWORK","_number":2,"created":"2020-03-09 11:17:17.000000000","uploader":{"_account_id":28948,"name":"Liang Fang","email":"liang.a.fang@intel.com","username":"liang"},"ref":"refs/changes/78/711878/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/711878/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/711878/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/711878/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/711878/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/711878/2"}}},"commit":{"parents":[{"commit":"3897c6f2b73a54b8dfd52ab5411159ca62c578f4","subject":"Move lodgeit to opendev tenant","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3897c6f2b73a54b8dfd52ab5411159ca62c578f4"}]}],"author":{"name":"Liang Fang","email":"liang.a.fang@intel.com","date":"2020-03-09 07:09:42.000000000","tz":480},"committer":{"name":"Liang Fang","email":"liang.a.fang@intel.com","date":"2020-03-09 11:00:12.000000000","tz":480},"subject":"New repo: devstack-plugin-open-cas","message":"New repo: devstack-plugin-open-cas\n\nThis repo is to add plugin to configure open-cas which is needed in\nvolume local cache feature.\n\nopen-cas: https://open-cas.github.io/\nNova Spec: https://review.opendev.org/#/c/689070/\nCinder Spec: https://review.opendev.org/#/c/684556/\n\nNeeded-By: https://review.opendev.org/#/c/711902/\nChange-Id: I176dc57afcfd627ad470ed8c6cf7fbb7299cbb35\nSigned-off-by: Liang Fang \u003cliang.a.fang@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a26717c6343386e6547621711e2d1e0c7f746e43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a26717c6343386e6547621711e2d1e0c7f746e43"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
