)]}'
{"id":"openstack%2Fproject-config~711903","triplet_id":"openstack%2Fproject-config~master~If100067962efcd1d472f238f8e6e53b1678b7e11","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"If100067962efcd1d472f238f8e6e53b1678b7e11","subject":"Setting OpenEdge Provider to 10 test nodes","status":"MERGED","created":"2020-03-09 11:11:50.000000000","updated":"2020-03-09 11:37:42.000000000","submitted":"2020-03-09 11:37:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711903-1583753862754-4545ba28","meta_rev_id":"93b7555df5ecd40fc6fe9e2ec7c6f2097ad4d948","_number":711903,"virtual_id_number":711903,"owner":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-09 11:37:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2020-03-09 11:27:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2020-03-09 11:27:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-09 11:27:31.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-03-09 11:37:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f3c0b27f32e5e6e42fe9bfd3a18be7eb0b6b31bb","author":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"date":"2020-03-09 11:11:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1d1ff921e5df84ac34b932bb81abd747b83898e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 11:19:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e75327aa30804a77b26f1ad7be9d6ecc : SUCCESS in 6m 27s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/b8cdf7a3d2e747b1b8c429ed52e49ec2 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"1c8e099466f5447904afb95d2912f922b6e9986a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-03-09 11:27:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nFirst test runs look fine, fast-approving to proceed with testing.","accounts_in_message":[],"_revision_number":1},{"id":"75899d59f3c217402f15e729edcb9cd099e98b12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 11:27:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"54eca4e6f247809be87b20cf4bce43af1be66c25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 11:37:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"93b7555df5ecd40fc6fe9e2ec7c6f2097ad4d948","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 11:37:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e801e04a2c4d4b2da01af0bd0a9f1ca9 : SUCCESS in 9m 38s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/9dd0f9a6a3f54718b3c24157a3490cf8 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e6bc951dc8d97d2cb56c00d4bc2e8a76dfa84bad","revisions":{"e6bc951dc8d97d2cb56c00d4bc2e8a76dfa84bad":{"kind":"REWORK","_number":1,"created":"2020-03-09 11:11:50.000000000","uploader":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"ref":"refs/changes/03/711903/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/03/711903/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/711903/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/711903/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/03/711903/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/03/711903/1"}}},"commit":{"parents":[{"commit":"9663df0c0164f9e9197c904e464cb1d42848d3ba","subject":"Start minimally using Open Edge environment","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9663df0c0164f9e9197c904e464cb1d42848d3ba"}]}],"author":{"name":"Donny Davis","email":"donny@fortnebula.com","date":"2020-03-09 11:10:58.000000000","tz":-240},"committer":{"name":"Donny Davis","email":"donny@fortnebula.com","date":"2020-03-09 11:11:49.000000000","tz":-240},"subject":"Setting OpenEdge Provider to 10 test nodes","message":"Setting OpenEdge Provider to 10 test nodes\n\nChange-Id: If100067962efcd1d472f238f8e6e53b1678b7e11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e6bc951dc8d97d2cb56c00d4bc2e8a76dfa84bad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e6bc951dc8d97d2cb56c00d4bc2e8a76dfa84bad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
