)]}'
{"id":"openstack%2Fproject-config~711910","triplet_id":"openstack%2Fproject-config~master~I149801b88484258fa05dabd1bc6a471462199e4c","project":"openstack/project-config","branch":"master","topic":"tripleo/validation-split","hashtags":[],"change_id":"I149801b88484258fa05dabd1bc6a471462199e4c","subject":"Add new Validation Framework projects","status":"MERGED","created":"2020-03-09 12:32:09.000000000","updated":"2020-03-10 18:38:26.000000000","submitted":"2020-03-10 18:38:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711910-1583865506833-9c168db8","meta_rev_id":"85f5ce5b8864bdc2c35b77733cf184377757b352","_number":711910,"virtual_id_number":711910,"owner":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-10 18:38:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-03-10 16:28:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2020-03-10 14:55:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":2,"date":"2020-03-10 18:26:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":1,"date":"2020-03-10 15:12:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":1,"date":"2020-03-10 18:26:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-09 12:33:06.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-09 12:33:09.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-09 13:18:44.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2020-03-10 14:55:42.000000000","updated_by":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"reviewer":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"state":"REVIEWER"},{"updated":"2020-03-10 15:12:38.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-03-10 16:28:38.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-03-10 18:26:47.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-03-10 18:38:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bf01748ae9e431025eed249692efcc3f5f17e48c","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-03-09 12:32:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5f385669dcaa9773fc8e785b2277deba2aa9285","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 12:39:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3a61c430350247bc9a631871f13d574d : SUCCESS in 5m 48s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/22edc1d167a74de6a3d70a0d53ab7e06 : FAILURE in 4m 30s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/cd5e410e91804eb49eeb13e54c3b395a : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b67e7c02e74b416a876da0f40992836b97ec6596","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-03-09 13:00:26.000000000","message":"Patch Set 1:\n\nrecheck\n\ndropped zuul.d directories from remote repositories.","accounts_in_message":[],"_revision_number":1},{"id":"ce7629be57d516bea47c40ef0a7bee6fd03cf9a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 13:09:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/127087318c4945db833f3eb7c7057f5b : SUCCESS in 7m 06s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/376ff5ac82534fc6a6582cecaaabd82d : FAILURE in 3m 35s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/8584149943ec4071975e8121d39130e3 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":1},{"id":"6fe693fd2613c0851c39033c6c9f90e0a1ebbe47","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-03-09 13:11:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1b1365338ba1c0d787a1bd3870007134c15d3f1c","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2020-03-09 13:18:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"28d1446f5aa50db7cf8c68be7dc55044101042bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 13:28:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f7fe464b26b041f0b7f22f32b3332d7d : SUCCESS in 10m 16s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f1ad2525b1a04b81895835c9a5387022 : SUCCESS in 4m 17s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/c3616381ee8a4ff5bb2baa9e1d6e593e : SUCCESS in 7m 51s","accounts_in_message":[],"_revision_number":2},{"id":"4ce99c80cb3e25620de05dd92bc5655f090edb8b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-09 13:31:51.000000000","message":"Patch Set 2: Code-Review-1\n\nThis needs a governance change linked via Needed-By, see https://docs.openstack.org/infra/manual/creators.html#add-new-repository-to-the-governance-repository","accounts_in_message":[],"_revision_number":2},{"id":"b3cf16847febf53c0c6835fba443be315be96cd2","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-03-09 13:55:31.000000000","message":"Patch Set 2: Workflow-1\n\nstill WIP - this is a topic for tomorrow meeting on #tripleo","accounts_in_message":[],"_revision_number":2},{"id":"8b8dbe192b96f5dd977e591b4c3c599fc938fcda","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2020-03-10 14:17:40.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"03707e88f7fe9144afd457f93462869fdbd029bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 14:31:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/dbe886faecbd4ef6ba6c8dacec1c286e : SUCCESS in 8m 49s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ba97b6eabe3042588999cb33d70705dd : SUCCESS in 4m 50s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/499b07a05f5441bc8c7046657b9ca726 : SUCCESS in 9m 48s","accounts_in_message":[],"_revision_number":3},{"id":"d500833714453e2364dd0f4ca75330c56a9fcc14","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2020-03-10 14:55:42.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c9b142a15b5907a685f3042214e9230b118cd3db","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-03-10 15:12:38.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"af8b15180f38bad6398e8f8fefe0ee1e880c27d0","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-03-10 16:28:38.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"cc5aa2e966a32c385ffe53445695e5fa63f634f4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-10 18:26:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"074ddb9590d76990a49d22ca44d1b8f4e5242c00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 18:27:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"af866bf394959153f6af8567c801f4de609b3a7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 18:38:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"be572572113d5d1da5b5a3ea6103df46ebbc1181","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 18:38:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/025c0b1950f148d7ab9a47fa65bb146a : SUCCESS in 8m 41s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e802d566494744ab9cf5c8f8d6356b4b : SUCCESS in 6m 40s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/7fd5dfa5808241f4b5e7e4d5117ba17b : SUCCESS in 9m 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f87fdbdc130d7c262d13a6bd06a2bce02581b3c7","revisions":{"7f6cabbf6caa5568c32da7e1c290a1779cdf4790":{"kind":"REWORK","_number":1,"created":"2020-03-09 12:32:09.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/10/711910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/711910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/711910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/711910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/711910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/711910/1"}}},"commit":{"parents":[{"commit":"d2ed9a0df7ee7484b2f765048de874147cb40bd6","subject":"Fix stackviz periodic job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d2ed9a0df7ee7484b2f765048de874147cb40bd6"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-02-27 14:26:36.000000000","tz":60},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-03-09 12:31:51.000000000","tz":60},"subject":"Add new Validation Framework projects","message":"Add new Validation Framework projects\n\nCreate two repositories for the Validation Framework:\n- validations-libs\n- validations-common\n\nThis is a first, needed step in order to get proper separation for the\ncode and libraries.\nThe tripleo-validations repository stay as is for now.\n\nChange-Id: I149801b88484258fa05dabd1bc6a471462199e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7f6cabbf6caa5568c32da7e1c290a1779cdf4790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7f6cabbf6caa5568c32da7e1c290a1779cdf4790"}]},"branch":"refs/heads/master"},"0c8fffe60321f0f5494226aae2a0859988597828":{"kind":"REWORK","_number":2,"created":"2020-03-09 13:11:30.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/10/711910/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/711910/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/711910/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/711910/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/711910/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/711910/2"}}},"commit":{"parents":[{"commit":"d2ed9a0df7ee7484b2f765048de874147cb40bd6","subject":"Fix stackviz periodic job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d2ed9a0df7ee7484b2f765048de874147cb40bd6"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-02-27 14:26:36.000000000","tz":60},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-03-09 13:11:13.000000000","tz":60},"subject":"Add new Validation Framework projects","message":"Add new Validation Framework projects\n\nCreate two repositories for the Validation Framework:\n- validations-libs\n- validations-common\n\nThis is a first, needed step in order to get proper separation for the\ncode and libraries.\nThe tripleo-validations repository stay as is for now.\n\nChange-Id: I149801b88484258fa05dabd1bc6a471462199e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0c8fffe60321f0f5494226aae2a0859988597828"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0c8fffe60321f0f5494226aae2a0859988597828"}]},"branch":"refs/heads/master"},"f87fdbdc130d7c262d13a6bd06a2bce02581b3c7":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-03-10 14:17:40.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/10/711910/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/711910/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/711910/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/711910/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/711910/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/711910/3"}}},"commit":{"parents":[{"commit":"d2ed9a0df7ee7484b2f765048de874147cb40bd6","subject":"Fix stackviz periodic job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d2ed9a0df7ee7484b2f765048de874147cb40bd6"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-02-27 14:26:36.000000000","tz":60},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2020-03-10 14:17:10.000000000","tz":60},"subject":"Add new Validation Framework projects","message":"Add new Validation Framework projects\n\nCreate two repositories for the Validation Framework:\n- validations-libs\n- validations-common\n\nThis is a first, needed step in order to get proper separation for the\ncode and libraries.\nThe tripleo-validations repository stay as is for now.\n\nNeeded-By: I545338f21b0ef308f524e585c44801eab8d3cac6\nChange-Id: I149801b88484258fa05dabd1bc6a471462199e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f87fdbdc130d7c262d13a6bd06a2bce02581b3c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f87fdbdc130d7c262d13a6bd06a2bce02581b3c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
