)]}'
{"id":"openstack%2Fproject-config~711914","triplet_id":"openstack%2Fproject-config~master~I9ab3608a0994f6e6fb049ff399c5f5ebb992440e","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I9ab3608a0994f6e6fb049ff399c5f5ebb992440e","subject":"Bumping OpenEdge test node commit to 20","status":"MERGED","created":"2020-03-09 12:53:42.000000000","updated":"2020-03-09 13:37:23.000000000","submitted":"2020-03-09 13:37:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711914-1583761043857-72e87165","meta_rev_id":"483abcf3cf7d355725f74192881ccf615f22a050","_number":711914,"virtual_id_number":711914,"owner":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-09 13:37:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-03-09 13:30:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2020-03-09 13:30:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-09 13:30:07.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-03-09 13:37:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d2f4cb2ad619dbb59ac890fe3189f4c257118cde","author":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"date":"2020-03-09 12:53:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eba26ab687b22fa73a6678fcbde9b9ac59329515","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 13:04:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/cea74895a9e14211859e4972220e13e1 : SUCCESS in 6m 21s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/c06c225bf6544679965f943d8e720759 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"983e1999ad1d7d9d97db252a0fdcdeac130ba573","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-09 13:30:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nsingle core approving node increase","accounts_in_message":[],"_revision_number":1},{"id":"12c8ca8285eedd516ac7ae14349fb94c46bec07b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 13:30:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f3825f9c2c567492d76e305bcf95831cd374b810","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 13:37:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"483abcf3cf7d355725f74192881ccf615f22a050","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 13:37:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0b2da49562a341478e50097dd09560db : SUCCESS in 6m 43s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/0d64d4719b2a4184a4e474eacdeafa49 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d716ed99d02a862066c8b7899497eee50f5ecd3d","revisions":{"d716ed99d02a862066c8b7899497eee50f5ecd3d":{"kind":"REWORK","_number":1,"created":"2020-03-09 12:53:42.000000000","uploader":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"ref":"refs/changes/14/711914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/14/711914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/711914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/711914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/711914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/14/711914/1"}}},"commit":{"parents":[{"commit":"e6bc951dc8d97d2cb56c00d4bc2e8a76dfa84bad","subject":"Setting OpenEdge Provider to 10 test nodes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e6bc951dc8d97d2cb56c00d4bc2e8a76dfa84bad"}]}],"author":{"name":"Donny Davis","email":"donny@fortnebula.com","date":"2020-03-09 12:52:53.000000000","tz":-240},"committer":{"name":"Donny Davis","email":"donny@fortnebula.com","date":"2020-03-09 12:52:53.000000000","tz":-240},"subject":"Bumping OpenEdge test node commit to 20","message":"Bumping OpenEdge test node commit to 20\n\nChange-Id: I9ab3608a0994f6e6fb049ff399c5f5ebb992440e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d716ed99d02a862066c8b7899497eee50f5ecd3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d716ed99d02a862066c8b7899497eee50f5ecd3d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
