)]}'
{"id":"openstack%2Fproject-config~711973","triplet_id":"openstack%2Fproject-config~master~I543222402a8e4d08cf075b2ed7f7c45d705799ca","project":"openstack/project-config","branch":"master","topic":"fix-partition-config","hashtags":[],"change_id":"I543222402a8e4d08cf075b2ed7f7c45d705799ca","subject":"Fix debian-buster partition config","status":"MERGED","created":"2020-03-09 19:24:38.000000000","updated":"2020-03-09 19:40:49.000000000","submitted":"2020-03-09 19:40:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711973-1583782849656-eb552905","meta_rev_id":"c93a0ef064c668116654705fb64b28fe601e6ea2","_number":711973,"virtual_id_number":711973,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-09 19:40:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-03-09 19:27:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-09 19:31:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-09 19:31:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-09 19:27:06.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-03-09 19:31:56.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-03-09 19:40:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b9f8db73ad02503fd72828e9a26635d3673d5ca","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-03-09 19:24:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"243a5f890e5cdcd0388f65ac6a115d997a039981","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-09 19:27:06.000000000","message":"Patch Set 1: Code-Review+2\n\nindent looks better...","accounts_in_message":[],"_revision_number":1},{"id":"99d9b6589d2227c8e46fdd6945de3b22acda22cd","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-03-09 19:31:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLooks like this has been broken since Ied44590c6260691a738ed105f8da0f4765b6c162 added the filesystem options ~6 weeks ago.","accounts_in_message":[],"_revision_number":1},{"id":"60ff7ad8938ecbdc9ace1be9bcd285d38399f321","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 19:33:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c71b50dee77d465496ffea4d8ae56770 : SUCCESS in 7m 53s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/6b95c98db7834ba1b1f9e8c1e83b6465 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"d3f09f05e4e36fa902e91e37d6e56d90b436ee4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 19:33:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"81f3bbd201d3ca76aff214d37b8f93adb9922721","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 19:40:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4820be58b17446779d804a64c43a0c5f : SUCCESS in 6m 40s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/36548cf88900407f9716e6ed7286ffc0 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"0dad32c712a2fe019a1bc841b41349b22f98e232","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 19:40:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4df89b9be5c043b5ec7555af863f656c60936c45","revisions":{"4df89b9be5c043b5ec7555af863f656c60936c45":{"kind":"REWORK","_number":1,"created":"2020-03-09 19:24:38.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/73/711973/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/711973/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/711973/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/711973/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/711973/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/711973/1"}}},"commit":{"parents":[{"commit":"d716ed99d02a862066c8b7899497eee50f5ecd3d","subject":"Bumping OpenEdge test node commit to 20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d716ed99d02a862066c8b7899497eee50f5ecd3d"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-03-09 19:21:27.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-03-09 19:21:27.000000000","tz":-420},"subject":"Fix debian-buster partition config","message":"Fix debian-buster partition config\n\nThere was a lack of indentation which led to:\n\n  Traceback (most recent call last):\n    File \"/usr/local/bin/dib-block-device\", line 11, in \u003cmodule\u003e\n      sys.exit(main())\n    File \"/usr/local/lib/python3.5/dist-packages/diskimage_builder/block_device/cmd.py\", line 120, in main\n      return bdc.main()\n    File \"/usr/local/lib/python3.5/dist-packages/diskimage_builder/block_device/cmd.py\", line 115, in main\n      self.args.func()\n    File \"/usr/local/lib/python3.5/dist-packages/diskimage_builder/block_device/cmd.py\", line 30, in cmd_init\n      self.bd.cmd_init()\n    File \"/usr/local/lib/python3.5/dist-packages/diskimage_builder/block_device/blockdevice.py\", line 260, in cmd_init\n      self.config \u003d config_tree_to_graph(self.config)\n    File \"/usr/local/lib/python3.5/dist-packages/diskimage_builder/block_device/config.py\", line 134, in config_tree_to_graph\n      \"Config entry top-level should be a single dict: %s\" % entry)\n  diskimage_builder.block_device.exception.BlockDeviceSetupException:\n    Config entry top-level should be a single dict:\n      {\n        \u0027base\u0027: \u0027image0\u0027,\n        \u0027partitioning\u0027: None,\n        \u0027partitions\u0027: [{\u0027flags\u0027: [\u0027boot\u0027, \u0027primary\u0027],\n          \u0027mkfs\u0027: {\u0027mount\u0027: {\u0027fstab\u0027: {\u0027options\u0027: \u0027defaults,nobarrier,noatime\u0027, \u0027fsck-passno\u0027: 1}, \u0027mount_point\u0027: \u0027/\u0027}},\n          \u0027size\u0027: \u0027100%\u0027, \u0027name\u0027: \u0027root\u0027}],\n        \u0027label\u0027: \u0027mbr\u0027}\n\nNote the paritioning value is None rather than a dict with those keys in\nit.\n\nChange-Id: I543222402a8e4d08cf075b2ed7f7c45d705799ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4df89b9be5c043b5ec7555af863f656c60936c45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4df89b9be5c043b5ec7555af863f656c60936c45"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
