)]}'
{"id":"openstack%2Fproject-config~712836","triplet_id":"openstack%2Fproject-config~master~I2ccb6aaad5866561fd6ec4118c718ee131c3a5f0","project":"openstack/project-config","branch":"master","topic":"nb-bionic","hashtags":[],"change_id":"I2ccb6aaad5866561fd6ec4118c718ee131c3a5f0","subject":"Revert \"Move fedora-30 builds to nb01.opendev.org\"","status":"MERGED","created":"2020-03-13 05:28:43.000000000","updated":"2020-03-13 06:33:06.000000000","submitted":"2020-03-13 06:33:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712836-1584081186849-cb471cbf","meta_rev_id":"06e154141de139d5b910ae00b0cd518283b6e7e4","_number":712836,"virtual_id_number":712836,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-13 06:33:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-13 05:30:39.000000000","_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-03-13 05:33:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-03-13 06:20:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-03-13 06:20:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 05:33:52.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-03-13 06:20:23.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-03-13 06:33:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12200d5a609d5b71086cd2adc195f88772d9277d","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-03-13 05:28:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a042babb9c61e10784aa75bc6665ca9507b5318","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-03-13 05:29:21.000000000","message":"Patch Set 1:\n\nnote, nb01.opendev.org container currently disabled, and the host is in the emergency file to stop it coming back.","accounts_in_message":[],"_revision_number":1},{"id":"57e7eef670654e27a2a1e7e04842d972f37f83bc","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-03-13 05:30:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7ea5412d76a5920fc8996b27314006646e5f4835","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-13 05:33:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5bc45b637bc5833a26992f36658bd09547a6d7de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 05:36:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5eb86ef9335a439b8b3d7e08a75dea7d : SUCCESS in 6m 19s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/de6dcbb62a0d4255b08f19d3648dfd85 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"50fdd45639a393edef28449ccc519913e48c27f5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-03-13 06:20:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a728faf15a833814877aa4dc07eb2f6706007ac9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 06:20:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"471a52f935bdd81374d9b4a435835a2f1d9e0af7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 06:33:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fd9010ff69843e74937a20e89c9165e8485f6df0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 06:33:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/89cd1b3adcc04b398f468d88f967680a : SUCCESS in 7m 25s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/bb2b2471baf8467888e488b9b31bf7da : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"98ba436966cea7fc9eff7fb8280043e3af6fd617","revisions":{"98ba436966cea7fc9eff7fb8280043e3af6fd617":{"kind":"REWORK","_number":1,"created":"2020-03-13 05:28:43.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/36/712836/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/36/712836/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/712836/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/712836/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/712836/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/36/712836/1"}}},"commit":{"parents":[{"commit":"9d7592df75a84cac484457e62dce917af80b092a","subject":"Move fedora-30 builds to nb01.opendev.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9d7592df75a84cac484457e62dce917af80b092a"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-03-13 05:22:08.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-03-13 05:25:41.000000000","tz":660},"subject":"Revert \"Move fedora-30 builds to nb01.opendev.org\"","message":"Revert \"Move fedora-30 builds to nb01.opendev.org\"\n\nThis reverts commit 9d7592df75a84cac484457e62dce917af80b092a which was\nmeant to make nb01.opendev.org the only builder of fedora-30, but did\nnot quite work as expected.  This replicated what we do on the ARM64\nbuilds; but what I missed was that the ARM builds have exclusive use\nof the ARM clouds -- in this case the new nb01.opendev.org thought\nthat all the other images in the existing clouds were foreign and\ntried to delete them.\n\nTake a different approach; replicate the nodepool.yaml file to a\nnb01.opendev.org specific one, but pause everything *but* the\nfedora-30 builds.  This should make the builder leave everything else\nalone, and just try building Fedora, which is what we want for the\ninitial POC.\n\nChange-Id: I2ccb6aaad5866561fd6ec4118c718ee131c3a5f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/98ba436966cea7fc9eff7fb8280043e3af6fd617"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/98ba436966cea7fc9eff7fb8280043e3af6fd617"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
