)]}'
{"id":"openstack%2Fproject-config~713575","triplet_id":"openstack%2Fproject-config~master~I7fce3a616ce52dce3bd3639bad37124a8bb1909b","project":"openstack/project-config","branch":"master","topic":"nb04","hashtags":[],"change_id":"I7fce3a616ce52dce3bd3639bad37124a8bb1909b","subject":"Add config for nb04.opendev.org","status":"MERGED","created":"2020-03-18 06:26:28.000000000","updated":"2020-03-18 20:56:52.000000000","submitted":"2020-03-18 20:56:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713575-1584565012813-e3550527","meta_rev_id":"76e7fd17582d14e285b76677e0510fce75c828b7","_number":713575,"virtual_id_number":713575,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-18 20:56:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2020-03-18 20:44:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-18 20:44:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-18 20:44:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-18 20:44:03.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2020-03-18 20:44:38.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-03-18 20:56:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f5c702d555de0d3a110ea8d56122b4213845fbcb","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-03-18 06:26:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3341d5b2b8e2ca943521b6de8602eb6ca149e9d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 07:19:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b51e298af25c47cfa46cc5df0ac53a8e : SUCCESS in 6m 12s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/2030e71e897b4a7ba6dc2fa003c48a0b : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":1},{"id":"4dda3021a2e469bd41c3c11204e50b7ab748fbc3","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-03-18 20:44:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"737d6019e6e84aee8fb36ae131bf9db36fe07ae1","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-03-18 20:44:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"be20ee1d80913d7fb984cf177497c1ccc84f3e56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 20:44:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6afdbdcddabddebb98c9cf5b8c169c79faec341b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 20:56:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"85066eeea5cd4f3166d5551cc655587d25d37a6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 20:56:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2bf20d89592045148d2153187c9ca58d : SUCCESS in 6m 34s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/f550251272434c5ea6b549024b79f3a5 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"616d91caa8bb12fd23d7a3cf1497190575c97f10","revisions":{"616d91caa8bb12fd23d7a3cf1497190575c97f10":{"kind":"REWORK","_number":1,"created":"2020-03-18 06:26:28.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/75/713575/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/75/713575/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/713575/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/713575/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/713575/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/75/713575/1"}}},"commit":{"parents":[{"commit":"4d09473b7fb1fa659c08f052323b62aa20724b6d","subject":"Merge \"Move diskimage-builder to #opendev\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4d09473b7fb1fa659c08f052323b62aa20724b6d"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-03-18 06:23:27.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-03-18 06:26:10.000000000","tz":660},"subject":"Add config for nb04.opendev.org","message":"Add config for nb04.opendev.org\n\nAdd a config for nb04.opendev.org that has *only* the Fedora image\nunpaused.  This is to be the new container-based builder, where we\nonly want it to build/manage this image, for now.\n\nChange-Id: I7fce3a616ce52dce3bd3639bad37124a8bb1909b\nStory: #2007407\nTask: #39064\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/616d91caa8bb12fd23d7a3cf1497190575c97f10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/616d91caa8bb12fd23d7a3cf1497190575c97f10"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
