)]}'
{"id":"openstack%2Fproject-config~713650","triplet_id":"openstack%2Fproject-config~master~Ib67fadae6d0adf1c5bcb64716b839b72671b779f","project":"openstack/project-config","branch":"master","topic":"new-project","hashtags":[],"change_id":"Ib67fadae6d0adf1c5bcb64716b839b72671b779f","subject":"Add Rook to StarlingX","status":"MERGED","created":"2020-03-18 13:32:39.000000000","updated":"2020-03-30 21:13:56.000000000","submitted":"2020-03-30 21:13:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713650-1585602836552-73754ddb","meta_rev_id":"affcb77402482daa1d1fa5ee97336da063a5aa74","_number":713650,"virtual_id_number":713650,"owner":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"date":"2020-03-30 05:46:44.000000000","_account_id":28602,"name":"chen haochuan","email":"haochuan.z.chen@intel.com","username":"haochuan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-30 21:13:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-23 18:56:32.000000000","_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-03-24 06:01:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-03-23 20:21:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":28602,"name":"chen haochuan","email":"haochuan.z.chen@intel.com","username":"haochuan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"},{"value":2,"date":"2020-03-30 20:43:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":28602,"name":"chen haochuan","email":"haochuan.z.chen@intel.com","username":"haochuan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"},{"value":1,"date":"2020-03-30 20:43:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"_account_id":28602,"name":"chen haochuan","email":"haochuan.z.chen@intel.com","username":"haochuan"},{"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-18 16:09:05.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-03-23 18:56:32.000000000","updated_by":{"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"},"reviewer":{"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"},"state":"REVIEWER"},{"updated":"2020-03-24 06:01:43.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-03-30 05:46:44.000000000","updated_by":{"_account_id":28602,"name":"chen haochuan","email":"haochuan.z.chen@intel.com","username":"haochuan"},"reviewer":{"_account_id":28602,"name":"chen haochuan","email":"haochuan.z.chen@intel.com","username":"haochuan"},"state":"REVIEWER"},{"updated":"2020-03-30 20:43:46.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-03-30 21:13:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"262e679344dd1708cc2bd6593cf2a892c57c0084","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2020-03-18 13:32:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0e9c7dc347dd0fd290c56aa7034e6c98bd593d9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-18 13:37:24.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"597da1e41045dd1f49de049fc1abb4c56c05d299","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 13:41:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7c2fee82b3634e11bcddfde8c713b221 : SUCCESS in 6m 30s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d0f491a431d148e790f1e554b40f9791 : FAILURE in 5m 54s","accounts_in_message":[],"_revision_number":1},{"id":"2fa1712b48239714913e47f26242455595f3fa5c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-03-18 16:09:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4eeb1975b51126a8fe873d45e756589cefe2504e","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2020-03-19 17:44:20.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d877acde643cc33850ffae543833ce9655a12594","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-19 18:18:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84998683cf491afdb40de3d9e3340524e5d7056b","author":{"_account_id":28602,"name":"chen haochuan","email":"haochuan.z.chen@intel.com","username":"haochuan"},"date":"2020-03-23 14:00:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d04ba858169562381991fb78b5f621ed33e43526","author":{"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"},"date":"2020-03-23 17:23:18.000000000","message":"Patch Set 1:\n\n\u003e (1 comment)\n\nIt\u0027s possible, all the starlingx projects are setup with pushMerge.","accounts_in_message":[],"_revision_number":1},{"id":"edea83b328b4ce8923e1c708432440355d4febaa","author":{"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"},"date":"2020-03-23 17:28:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5e2967fc2132e55bd268d233e9fa0df3cb47536b","author":{"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"},"date":"2020-03-23 17:28:49.000000000","message":"Patch Set 1:\n\nI removed the upstream reference as this should start from an empty repo.","accounts_in_message":[],"_revision_number":1},{"id":"30c3fc60b2ebf86b9f8af6e6d08e0bd39dc4dd18","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-23 17:31:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c7afbce70251a37b6de212549ccd33775b076e3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 17:41:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e0dacd103d52475ab99d53924d3e7da2 : SUCCESS in 8m 29s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/feb6932b8b4a4593a921b8f364caafb7 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":2},{"id":"36e330c42b6f95363a94c211fbfc9f5d3945ab38","author":{"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"},"date":"2020-03-23 17:48:22.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n\nAs I said we are following the same pattern as the rest of the StarlingX Acl files have. Thats why we would prefer to keep it in.","accounts_in_message":[],"_revision_number":2},{"id":"68083d98bf32ddee50cb12bc4c69990f81eb08f8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-23 17:51:11.000000000","message":"Patch Set 1:\n\n\u003e \u003e (1 comment)\n \u003e \n \u003e As I said we are following the same pattern as the rest of the\n \u003e StarlingX Acl files have. Thats why we would prefer to keep it in.\n\nPlease double check - not all starlingx files use it. So, what\u0027s your use case for it?","accounts_in_message":[],"_revision_number":1},{"id":"48db7a689afc2076aba7fca83197d3db76fb003e","author":{"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"},"date":"2020-03-23 18:56:32.000000000","message":"Patch Set 2:\n\n\u003e \u003e \u003e (1 comment)\n \u003e \u003e\n \u003e \u003e As I said we are following the same pattern as the rest of the\n \u003e \u003e StarlingX Acl files have. Thats why we would prefer to keep it\n \u003e in.\n \u003e \n \u003e Please double check - not all starlingx files use it. So, what\u0027s\n \u003e your use case for it?\nThe acl files that don\u0027t have pushMerge are textual (docs, governance, elections, and specs)  All code based project acl files ones have the pushMerge line.  The use case is feature branch and upgrades that can\u0027t be done on the master branch.  StarlingX does things a little differently than other OpenStack projects.","accounts_in_message":[],"_revision_number":2},{"id":"afcc12ef19c3476600d2c98af3315811650d8ba4","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2020-03-23 20:21:48.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3c9dc00c924959b75a4eb1ebdcbec4ba3dc42efb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-24 06:01:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7e66ab4021f81c120e2bc19c059ac7eae657f9b0","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-24 15:30:23.000000000","message":"Patch Set 2: Workflow-1\n\nSetting this to WIP since we currently cannot merge in such changes - we have a freeze on repo creation/changes while some changes are done to the infrastructure. Once the freeze is lifted, I\u0027ll un-WIP.","accounts_in_message":[],"_revision_number":2},{"id":"256ccec6243c21bcc5ab1028dbef88822e1fe9a3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-28 16:14:10.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"033a3737625fe1e61e0edf1167a07f389513002e","author":{"_account_id":28602,"name":"chen haochuan","email":"haochuan.z.chen@intel.com","username":"haochuan"},"date":"2020-03-30 04:49:27.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: -Workflow\n\nNew project will be created? Thanks!","accounts_in_message":[],"_revision_number":2},{"id":"3f53513e6d84d7d628d47c6abcd895b151dff336","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-30 05:44:10.000000000","message":"Patch Set 2:\n\nWe still have a few problems with project creation and will slowly move forward with new project approvals now.","accounts_in_message":[],"_revision_number":2},{"id":"fdf5342b4d93069f8caf45fc0acd5c6a6208cfd2","author":{"_account_id":28602,"name":"chen haochuan","email":"haochuan.z.chen@intel.com","username":"haochuan"},"date":"2020-03-30 05:46:44.000000000","message":"Patch Set 2:\n\n\u003e We still have a few problems with project creation and will slowly\n \u003e move forward with new project approvals now.\n\nthanks!","accounts_in_message":[],"_revision_number":2},{"id":"06f98a6c717d312e4edf99cb755892b579e9e19f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-03-30 18:09:15.000000000","message":"Topic changed from rook-ceph to new-project","accounts_in_message":[],"_revision_number":2},{"id":"92062dc54a149dfaa8987bce58f5513c08419000","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-03-30 20:43:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f13c0dd9942983e7d343ed5a3c27d26fde0fa463","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 20:44:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"abd8856428132d6eb2ae23a10ab510507f7db14d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 21:13:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"affcb77402482daa1d1fa5ee97336da063a5aa74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 21:13:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5867a0ce6c31468c8b97273045c94015 : SUCCESS in 6m 02s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d342a20a67564edab0aadef9c37dc99b : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8dd62c2069d75296be888a603c0d3f228ae472f9","revisions":{"44450029c0f0da50c7b2f22502250f7fdf12d458":{"kind":"REWORK","_number":1,"created":"2020-03-18 13:32:39.000000000","uploader":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"ref":"refs/changes/50/713650/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/713650/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/713650/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/713650/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/713650/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/713650/1"}}},"commit":{"parents":[{"commit":"57912b10c37365b367a040a04be29c404b286393","subject":"vexxhost: add openstack/openstack-zuul-jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/57912b10c37365b367a040a04be29c404b286393"}]}],"author":{"name":"Scott Little","email":"scott.little@windriver.com","date":"2020-03-18 13:31:47.000000000","tz":-240},"committer":{"name":"Scott Little","email":"scott.little@windriver.com","date":"2020-03-18 13:31:56.000000000","tz":-240},"subject":"Add Rook to StarlingX","message":"Add Rook to StarlingX\n\nChange-Id: Ib67fadae6d0adf1c5bcb64716b839b72671b779f\nSigned-off-by: Scott Little \u003cscott.little@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/44450029c0f0da50c7b2f22502250f7fdf12d458"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/44450029c0f0da50c7b2f22502250f7fdf12d458"}]},"branch":"refs/heads/master"},"8dd62c2069d75296be888a603c0d3f228ae472f9":{"kind":"REWORK","_number":2,"created":"2020-03-23 17:28:08.000000000","uploader":{"_account_id":28607,"name":"Saul Wold","email":"saul.wold@windriver.com","username":"sgw-starlingx"},"ref":"refs/changes/50/713650/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/50/713650/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/713650/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/713650/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/50/713650/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/50/713650/2"}}},"commit":{"parents":[{"commit":"57912b10c37365b367a040a04be29c404b286393","subject":"vexxhost: add openstack/openstack-zuul-jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/57912b10c37365b367a040a04be29c404b286393"}]}],"author":{"name":"Scott Little","email":"scott.little@windriver.com","date":"2020-03-18 13:31:47.000000000","tz":-240},"committer":{"name":"Saul Wold","email":"sgw@linux.intel.com","date":"2020-03-23 17:26:02.000000000","tz":-420},"subject":"Add Rook to StarlingX","message":"Add Rook to StarlingX\n\nChange-Id: Ib67fadae6d0adf1c5bcb64716b839b72671b779f\nSigned-off-by: Scott Little \u003cscott.little@windriver.com\u003e\nSigned-off-by: Saul Wold \u003csgw@linux.intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8dd62c2069d75296be888a603c0d3f228ae472f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8dd62c2069d75296be888a603c0d3f228ae472f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
