)]}'
{"id":"openstack%2Fproject-config~713815","triplet_id":"openstack%2Fproject-config~master~Ia800f77e5998a3f0fbe62ff6d027471d2a11bf0a","project":"openstack/project-config","branch":"master","topic":"cloudkitty-specs","hashtags":[],"change_id":"Ia800f77e5998a3f0fbe62ff6d027471d2a11bf0a","subject":"Fix cloudkitty-specs on specs.o.o","status":"MERGED","created":"2020-03-19 10:10:59.000000000","updated":"2020-03-19 11:14:10.000000000","submitted":"2020-03-19 11:11:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713815-1584616315810-4d5b2a18","meta_rev_id":"af898abcc164106fa3ddc338b045d46b1feab397","_number":713815,"virtual_id_number":713815,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-19 11:11:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2020-03-19 10:59:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2020-03-19 10:59:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-19 10:59:57.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-03-19 11:11:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c9d403fb1984b1414581466ece5fe89b40c148ed","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-19 10:10:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47e6359568b464a00bce2724c41c88dc6d4e08fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 10:23:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/23aff08fae974784881ea8e00feee1ee : SUCCESS in 4m 55s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/35b876d1c4694f11a6242f477104d8dd : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":1},{"id":"cef495c67948174c353f1f3212ef2b26b870c436","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-03-19 10:59:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nSingle approving obvious fix","accounts_in_message":[],"_revision_number":1},{"id":"d16e41e981cf6e1ca8e3d3c53997f9cba1cdb0a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 11:00:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06482278d2a54330b17fb7c1ce79efd32aa30981","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 11:11:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/78ca975c21404b25be4a2430a6bb1a19 : SUCCESS in 3m 47s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/25e520c1e46241fd8b3074060fc3f02f : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":1},{"id":"d4d8d5ab2e66fe9b9e3eab5acf26d126291e377d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 11:11:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"87d334af8a7921ef28b0f15e385dade5c92f51f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 11:14:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs-site https://zuul.opendev.org/t/openstack/build/34acbd126b164ef3b54bf47007b88d26 : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"df25a0d5a147ebd3c1b3cbb1013501e86bc3b4ec","revisions":{"df25a0d5a147ebd3c1b3cbb1013501e86bc3b4ec":{"kind":"REWORK","_number":1,"created":"2020-03-19 10:10:59.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/15/713815/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/15/713815/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/713815/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/713815/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/713815/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/15/713815/1"}}},"commit":{"parents":[{"commit":"3b443133917ef59e5844aebd32450954330480e5","subject":"Merge \"Add publish-to-pypi template for ovn-octavia-provider\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3b443133917ef59e5844aebd32450954330480e5"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-19 10:10:03.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-19 10:10:03.000000000","tz":60},"subject":"Fix cloudkitty-specs on specs.o.o","message":"Fix cloudkitty-specs on specs.o.o\n\nThe repo is named cloudkitty-specs, not cloudkitty-specifications. Fix\nname so that central index page works.\n\nChange-Id: Ia800f77e5998a3f0fbe62ff6d027471d2a11bf0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/df25a0d5a147ebd3c1b3cbb1013501e86bc3b4ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/df25a0d5a147ebd3c1b3cbb1013501e86bc3b4ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
