)]}'
{"id":"openstack%2Fproject-config~713925","triplet_id":"openstack%2Fproject-config~master~I46bf9e9527941f2120de59fdbc355337f4c019c1","project":"openstack/project-config","branch":"master","topic":"rename-infra-manual","hashtags":[],"change_id":"I46bf9e9527941f2120de59fdbc355337f4c019c1","subject":"Update infra-manual links","status":"MERGED","created":"2020-03-19 16:28:52.000000000","updated":"2020-03-21 17:36:50.000000000","submitted":"2020-03-21 17:33:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713925-1584812005338-577aa814","meta_rev_id":"ee2bfbfc9fd7d86d498a8750d9c5a95a69262395","_number":713925,"virtual_id_number":713925,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"date":"2020-03-21 15:49:56.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-21 17:33:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"recommended":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"all":[{"value":1,"date":"2020-03-21 10:05:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":2,"date":"2020-03-21 17:23:32.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-21 15:45:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":1,"date":"2020-03-21 17:23:32.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-21 10:05:29.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"},{"updated":"2020-03-21 15:45:32.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-03-21 17:23:32.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-03-21 17:33:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cf09d528963664675056efbd55f7a4cee356437c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-19 16:28:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5535f0d557edcddb1846b3bc24f00f5f0f458f2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 16:29:33.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"95bafb1ff64a96c6419ac495852e282cea4b0a3b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-19 16:38:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"580ab14a0e9e7a7d3abede040c8ad2aa31a7edb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 16:38:58.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":2},{"id":"840949b74395820cac678381b86fe1249971acfc","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-19 16:39:25.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"0aa586e1ec555f4a21998f11b61d64d4bfd6f8c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 16:41:08.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":3},{"id":"3206f8a16f4cdefb44ab2f218be50a472b1473a7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-20 16:03:03.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"753f1f5b0176997c6716e3b1a635c3308cca3848","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-20 16:03:16.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"a55eb53a73a7f65e3fe9f68fae732ff828ab7493","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 16:21:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/9de4a30793744f90a5786dc00078eaa9 : SUCCESS in 5m 43s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e9e779f7f1934e18852fcf00c2e07b67 : SUCCESS in 9m 03s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/97d99c8e440f48b68b03c89f9cebe79b : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":4},{"id":"6252374c1873f4f7bc57586e3a70e0f60a61864a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-20 19:43:26.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"346ecc7cdc1cb78e41627c5f70159b00cb1e33cb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-20 20:42:07.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8a29bdad923c912725f5ece92be8baa54b436fda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 20:50:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/82b2a00a65464d1a906acae4e73d2527 : SUCCESS in 3m 20s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c458d92e7ebc45759666e1bd5a45f5d8 : SUCCESS in 6m 47s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/a9547ebcc10b4c95a24bcecad785b350 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":5},{"id":"f8bb9c8d485205362e8e894a48ee07481109860c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-20 21:08:38.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"165e83a6c0c1434c0bfaf339f78ba4164d959181","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 21:19:45.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/9009c05732ec4a8eb242c3d1862ec0f7 : SUCCESS in 3m 42s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/eedb2e64fab941569ccc4a56adab0f1d : SUCCESS in 6m 21s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/edae8d5510a940f3b815087d3bb81b81 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":6},{"id":"9f1982cf24bc383e811c81c6a9429662e5416651","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-21 08:29:50.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"be3e99f736de92bf62112a6b4d33d7b600400af7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 08:36:35.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/cf9d97996a394a688b71e8e0efa80697 : SUCCESS in 2m 54s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f788493cf604458aad07139e6bd2dc42 : SUCCESS in 6m 01s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/966edf9ccf4b4efaa15c08de9c4056e4 : SUCCESS in 4m 16s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/8a350b6ae41e477e927454d9914b60f7 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":7},{"id":"002e142545cd46e2a605328d9a349b801b260ce1","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-03-21 10:05:29.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"41fe1ebdf0c08a23cfa31389ae25e6e4834b4f62","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-03-21 15:45:32.000000000","message":"Patch Set 7: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"055cd859811892a3c82a3213424493897fc1d590","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-21 15:49:56.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"754f5048d8226d5df11b6be69142ff5859c72383","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-21 17:23:32.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"b66772bbb761a163a10fbaafdc1c1179c65bc094","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 17:23:55.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"2d975beaff093a2ba3f1614739f01be8942ae381","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 17:33:24.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/8fffa7b1afcc4df4a084fe1b194b679d : SUCCESS in 3m 19s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2ebc1fdeb6674a53920888ebea1b61a3 : SUCCESS in 5m 39s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/0e92b28df98849d3903e828fb833daa4 : SUCCESS in 3m 23s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/1cbe797a9adc4cf89fbfb6c2db9f1d72 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":7},{"id":"cbfb471c2773e983a39bfa51f6fd5e3462a28b6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 17:33:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"9f4d4d0497bca72f7150e8cd08e892c66604ae0a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 17:36:50.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-infra-index https://zuul.opendev.org/t/openstack/build/3de88717965e4ba786d892ee5735dafc : SUCCESS in 1m 30s\n- promote-openstack-specs-site https://zuul.opendev.org/t/openstack/build/3d338009ad5a4fe9b15b25446df42628 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"a494d238b0d9e72193cb8dab92796a11e559aa51","revisions":{"a6f9a79aeb2ae40ae47188af1774bec09b60c865":{"kind":"REWORK","_number":1,"created":"2020-03-19 16:28:52.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/25/713925/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/25/713925/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/713925/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/713925/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/713925/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/25/713925/1"}}},"commit":{"parents":[{"commit":"22f46d6a71d204b02b8732e08146d201f4b32d29","subject":"Move openstack/infra-manuals to opendev","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/22f46d6a71d204b02b8732e08146d201f4b32d29"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-19 16:24:42.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-19 16:24:42.000000000","tz":60},"subject":"Update infra-manual links","message":"Update infra-manual links\n\nThe infra-manual publishes now to docs.opendev.org, change links to it.\n\nChange-Id: I46bf9e9527941f2120de59fdbc355337f4c019c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a6f9a79aeb2ae40ae47188af1774bec09b60c865"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a6f9a79aeb2ae40ae47188af1774bec09b60c865"}]},"branch":"refs/heads/master"},"9302a645c0c9324c9236429ef2a121acc61520f9":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2020-03-19 16:38:18.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/25/713925/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/25/713925/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/713925/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/713925/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/713925/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/25/713925/2"}}},"commit":{"parents":[{"commit":"9b79bead302710ff8e7e0a65429b330c8b65922f","subject":"Move openstack/infra-manuals to opendev","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9b79bead302710ff8e7e0a65429b330c8b65922f"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-19 16:24:42.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-19 16:37:59.000000000","tz":60},"subject":"Update infra-manual links","message":"Update infra-manual links\n\nThe infra-manual publishes now to docs.opendev.org, change links to it.\n\nDepends-On: https://review.opendev.org/713929\nChange-Id: I46bf9e9527941f2120de59fdbc355337f4c019c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9302a645c0c9324c9236429ef2a121acc61520f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9302a645c0c9324c9236429ef2a121acc61520f9"}]},"branch":"refs/heads/master"},"24f7b9a41c7eda8e72cc44fa71758b7b97b8e79d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-03-19 16:39:25.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/25/713925/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/25/713925/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/713925/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/713925/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/713925/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/25/713925/3"}}},"commit":{"parents":[{"commit":"9b79bead302710ff8e7e0a65429b330c8b65922f","subject":"Move openstack/infra-manuals to opendev","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9b79bead302710ff8e7e0a65429b330c8b65922f"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-19 16:24:42.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-19 16:38:50.000000000","tz":60},"subject":"Update infra-manual links","message":"Update infra-manual links\n\nThe infra-manual publishes now to docs.opendev.org, change links to it.\n\nDepends-On: https://review.opendev.org/713929\nNeeded-By: https://review.opendev.org/713930\nChange-Id: I46bf9e9527941f2120de59fdbc355337f4c019c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/24f7b9a41c7eda8e72cc44fa71758b7b97b8e79d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/24f7b9a41c7eda8e72cc44fa71758b7b97b8e79d"}]},"branch":"refs/heads/master"},"113c924ec24e5342084c1e68213f974631e77272":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-03-20 16:03:16.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/25/713925/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/25/713925/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/713925/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/713925/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/713925/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/25/713925/4"}}},"commit":{"parents":[{"commit":"11bd536edc698c9beecebe0d3ca365046024f235","subject":"Rename x/devstack-plugin-nfs to openstack/devstack-plugin-nfs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/11bd536edc698c9beecebe0d3ca365046024f235"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-19 16:24:42.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-03-20 16:03:16.000000000","tz":0},"subject":"Update infra-manual links","message":"Update infra-manual links\n\nThe infra-manual publishes now to docs.opendev.org, change links to it.\n\nDepends-On: https://review.opendev.org/713929\nNeeded-By: https://review.opendev.org/713930\nChange-Id: I46bf9e9527941f2120de59fdbc355337f4c019c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/113c924ec24e5342084c1e68213f974631e77272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/113c924ec24e5342084c1e68213f974631e77272"}]},"branch":"refs/heads/master"},"4c1898cab648dae28e55bb0693e38e2700c43063":{"kind":"REWORK","_number":5,"created":"2020-03-20 20:42:07.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/25/713925/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/25/713925/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/713925/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/713925/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/713925/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/25/713925/5"}}},"commit":{"parents":[{"commit":"11bd536edc698c9beecebe0d3ca365046024f235","subject":"Rename x/devstack-plugin-nfs to openstack/devstack-plugin-nfs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/11bd536edc698c9beecebe0d3ca365046024f235"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-19 16:24:42.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-20 20:40:51.000000000","tz":60},"subject":"Update infra-manual links","message":"Update infra-manual links\n\nThe infra-manual publishes now to docs.opendev.org, change links to it.\n\nNew location is: https://docs.opendev.org/opendev/infra-manual/latest/\n\nDepends-On: https://review.opendev.org/713929\nNeeded-By: https://review.opendev.org/713930\nChange-Id: I46bf9e9527941f2120de59fdbc355337f4c019c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4c1898cab648dae28e55bb0693e38e2700c43063"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4c1898cab648dae28e55bb0693e38e2700c43063"}]},"branch":"refs/heads/master"},"07f8c328f211a89dda1507e4be28712c5327756f":{"kind":"REWORK","_number":6,"created":"2020-03-20 21:08:38.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/25/713925/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/25/713925/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/713925/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/713925/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/713925/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/25/713925/6"}}},"commit":{"parents":[{"commit":"11bd536edc698c9beecebe0d3ca365046024f235","subject":"Rename x/devstack-plugin-nfs to openstack/devstack-plugin-nfs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/11bd536edc698c9beecebe0d3ca365046024f235"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-19 16:24:42.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-20 21:08:32.000000000","tz":60},"subject":"Update infra-manual links","message":"Update infra-manual links\n\nThe infra-manual publishes now to docs.opendev.org, change links to it.\n\nNew location is: https://docs.opendev.org/opendev/infra-manual/latest/\n\nDepends-On: https://review.opendev.org/713929\nNeeded-By: https://review.opendev.org/713930\nChange-Id: I46bf9e9527941f2120de59fdbc355337f4c019c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/07f8c328f211a89dda1507e4be28712c5327756f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/07f8c328f211a89dda1507e4be28712c5327756f"}]},"branch":"refs/heads/master"},"a494d238b0d9e72193cb8dab92796a11e559aa51":{"kind":"REWORK","_number":7,"created":"2020-03-21 08:29:50.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/25/713925/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/25/713925/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/713925/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/713925/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/713925/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/25/713925/7"}}},"commit":{"parents":[{"commit":"11bd536edc698c9beecebe0d3ca365046024f235","subject":"Rename x/devstack-plugin-nfs to openstack/devstack-plugin-nfs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/11bd536edc698c9beecebe0d3ca365046024f235"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-19 16:24:42.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-21 08:29:14.000000000","tz":60},"subject":"Update infra-manual links","message":"Update infra-manual links\n\nThe infra-manual publishes now to docs.opendev.org, change links to it.\n\nNew location is: https://docs.opendev.org/opendev/infra-manual/latest/\n\nLink specs.o.o contributors link to OpenStack Contributor Guide.\n\nChange http to https for docs.o.o and specs.o.o everywhere.\n\nDepends-On: https://review.opendev.org/713929\nNeeded-By: https://review.opendev.org/713930\nChange-Id: I46bf9e9527941f2120de59fdbc355337f4c019c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a494d238b0d9e72193cb8dab92796a11e559aa51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a494d238b0d9e72193cb8dab92796a11e559aa51"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
