)]}'
{"id":"openstack%2Fproject-config~714066","triplet_id":"openstack%2Fproject-config~master~I13fa07754e38281c63dcf0eceaa4c3b3c2715618","project":"openstack/project-config","branch":"master","topic":"fix-check-approval","hashtags":[],"change_id":"I13fa07754e38281c63dcf0eceaa4c3b3c2715618","subject":"check-release-approval: handle no-review case","status":"MERGED","created":"2020-03-20 10:27:21.000000000","updated":"2020-03-23 14:44:30.000000000","submitted":"2020-03-23 14:44:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714066-1584974670675-68d14c44","meta_rev_id":"627a401a03699e39fbd98be9c99d041bf84a07a2","_number":714066,"virtual_id_number":714066,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-03-20 13:07:31.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-23 14:44:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-03-20 12:47:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-23 14:27:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-03-20 13:16:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-23 14:27:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-20 12:47:31.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-20 13:16:45.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-03-23 14:27:21.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-03-23 14:44:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f7d67a43b1c0ab161d3fe89270cb2818e4605283","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-03-20 10:27:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d630eed60427fd8f5f0f8162f13cb2cd6a06c2f5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-20 10:34:22.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nNit inline","accounts_in_message":[],"_revision_number":1},{"id":"905f5dffb521fe7cf8615f8b28d58fadb6bd127a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 10:36:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1b51bd8068584e7e9af9375306884d2a : FAILURE in 3m 48s","accounts_in_message":[],"_revision_number":1},{"id":"767c543cfb13f605888538de5f70a481027991e4","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-03-20 10:38:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"090a4e39cc6a6c6492c3e32276ac589549289ded","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-03-20 10:40:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d0f4eda4315f8a24a82f3a6cb90d585a93440baa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 10:48:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2a3fc02c91ea4b98aa9237941e401f70 : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":3},{"id":"967ad0a0051d18747a772769500d4fa3284dce4c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-20 12:47:31.000000000","message":"Patch Set 3: Code-Review+1\n\nLooks good. From my understanding of the situation, this should handle it now on the initial push from PTL or liaison.","accounts_in_message":[],"_revision_number":3},{"id":"f4c043c336d65e5edd561a80d6be822b96c8338a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-03-20 13:07:31.000000000","message":"Patch Set 3:\n\nand on subsequent patchset pushes which reset votes","accounts_in_message":[],"_revision_number":3},{"id":"64de179c4eebfb3f03b4a7267c559982cd2e0fed","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-20 13:16:45.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5f307b4ab33f23f5e3e663068ca94cdc867179b7","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-03-23 14:27:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f960afe8e34d85f51f77786e52fdc720cca78e06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 14:27:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"49e6f1cad7be5a36747da61a83fdcdd903267b1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 14:44:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"06cf057073dbcfb172dccc6f418848f42c0059d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 14:44:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8dae557124d449a6954e1d209db1e03c : SUCCESS in 9m 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2ce6a551ac24d70a72b49bff573e40425b7fd663","revisions":{"bd00c3e7ba0806acc21905c7a57874b436b19a1d":{"kind":"REWORK","_number":1,"created":"2020-03-20 10:27:21.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/66/714066/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/66/714066/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/714066/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/714066/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/714066/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/66/714066/1"}}},"commit":{"parents":[{"commit":"d435ae141aaa0d5410819d4f991d7655ca900275","subject":"Revise gerrit ACLs for devstack-plugin-open-cas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d435ae141aaa0d5410819d4f991d7655ca900275"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-03-20 10:17:03.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-03-20 10:17:03.000000000","tz":60},"subject":"check-release-approval: handle no-review case","message":"check-release-approval: handle no-review case\n\nWhen no review is posted yet, Gerrit returns simplified \u0027labels\u0027\ndata. In particular it\u0027s missing the [\u0027labels\u0027][\u0027Code-Review][\u0027all\u0027]\ndictionary, which we assumed would always be present.\n\nWe should only add approvers from the reviews if the \u0027all\u0027 key is\nprovided, and otherwise just work from the owner email.\n\nRemove all changes pushed to investigate the issue: use a narrow\nquery again (rather than the /detail call) and no longer catch the\nexception.\n\nChange-Id: I13fa07754e38281c63dcf0eceaa4c3b3c2715618\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bd00c3e7ba0806acc21905c7a57874b436b19a1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bd00c3e7ba0806acc21905c7a57874b436b19a1d"}]},"branch":"refs/heads/master"},"df3175eb4e11fc2a667d57d5bdc5a8f76ad0adbe":{"kind":"REWORK","_number":2,"created":"2020-03-20 10:38:04.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/66/714066/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/66/714066/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/714066/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/714066/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/714066/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/66/714066/2"}}},"commit":{"parents":[{"commit":"d435ae141aaa0d5410819d4f991d7655ca900275","subject":"Revise gerrit ACLs for devstack-plugin-open-cas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d435ae141aaa0d5410819d4f991d7655ca900275"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-03-20 10:17:03.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-03-20 10:37:53.000000000","tz":60},"subject":"check-release-approval: handle no-review case","message":"check-release-approval: handle no-review case\n\nWhen no review is posted yet, Gerrit returns simplified \u0027labels\u0027\ndata. In particular it\u0027s missing the [\u0027labels\u0027][\u0027Code-Review][\u0027all\u0027]\ndictionary, which we assumed would always be present.\n\nWe should only add approvers from the reviews if the \u0027all\u0027 key is\nprovided, and otherwise just work from the owner email.\n\nRemove all changes pushed to investigate the issue: use a narrow\nquery again (rather than the /detail call) and no longer catch the\nexception.\n\nChange-Id: I13fa07754e38281c63dcf0eceaa4c3b3c2715618\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/df3175eb4e11fc2a667d57d5bdc5a8f76ad0adbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/df3175eb4e11fc2a667d57d5bdc5a8f76ad0adbe"}]},"branch":"refs/heads/master"},"2ce6a551ac24d70a72b49bff573e40425b7fd663":{"kind":"REWORK","_number":3,"created":"2020-03-20 10:40:14.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/66/714066/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/66/714066/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/714066/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/714066/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/714066/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/66/714066/3"}}},"commit":{"parents":[{"commit":"d435ae141aaa0d5410819d4f991d7655ca900275","subject":"Revise gerrit ACLs for devstack-plugin-open-cas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d435ae141aaa0d5410819d4f991d7655ca900275"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-03-20 10:17:03.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-03-20 10:39:44.000000000","tz":60},"subject":"check-release-approval: handle no-review case","message":"check-release-approval: handle no-review case\n\nWhen no review is posted yet, Gerrit returns simplified \u0027labels\u0027\ndata. In particular it\u0027s missing the [\u0027labels\u0027][\u0027Code-Review][\u0027all\u0027]\ndictionary, which we assumed would always be present.\n\nWe should only add approvers from the reviews if the \u0027all\u0027 key is\nprovided, and otherwise just work from the owner email.\n\nRemove all changes pushed to investigate the issue: use a narrow\nquery again (rather than the /detail call) and no longer catch the\nexception.\n\nChange-Id: I13fa07754e38281c63dcf0eceaa4c3b3c2715618\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2ce6a551ac24d70a72b49bff573e40425b7fd663"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2ce6a551ac24d70a72b49bff573e40425b7fd663"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
