)]}'
{"id":"openstack%2Fproject-config~714160","triplet_id":"openstack%2Fproject-config~master~I815c71fc7e924bd7fcfbe7851caad92d370da812","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I815c71fc7e924bd7fcfbe7851caad92d370da812","subject":"Don\u0027t load zuul objects from opendev/gerrit","status":"MERGED","created":"2020-03-20 16:05:55.000000000","updated":"2020-03-20 17:29:12.000000000","submitted":"2020-03-20 17:29:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714160-1584725352976-1d1e633c","meta_rev_id":"c3bbe72c7df7f9f3619d7674bd1d790811f8e04c","_number":714160,"virtual_id_number":714160,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-20 17:29:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2020-03-20 17:12:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2020-03-20 16:31:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2020-03-20 17:12:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-20 16:31:42.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-03-20 17:12:31.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2020-03-20 17:29:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"93669ab1f345a92a20a1319a81816cdd34578d7f","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-20 16:05:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8c5151ec7375c72d41369bcb86c4b0a10dad638","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-20 16:07:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3339b885a9a7960d665986ffc579a5adcd56d87d","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-20 16:12:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9b26539accf410dbd0e3fd28bc379e020d9eabb7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-20 16:19:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6399f8fd08b2c56085f3ededd34645dfbdd7c1ca","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-20 16:21:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3cdd96c6eaebd4de8908d774d70d3cbf347ffaa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 16:31:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ced7e16d4c4040ee900879e1a9d6fd79 : SUCCESS in 8m 17s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e4ca56ffa24946cb8337f537ca396610 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":3},{"id":"3d91c4d6c1b9df583c345419a0a6fea9df422320","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-20 16:31:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bd8a4e444eb7471dc16bf887257d9e156b4e7e93","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-03-20 17:12:31.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"049ee1fb9b7ccbf9add24a3ae0d263ae1ffc0eb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 17:12:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"61d7a0a6cd231e765157ac63db78db8e226f70ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 17:29:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c689de6dc02f45289f7a53851b128edc : SUCCESS in 8m 11s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d894722f79ab440386fc8731d5d2f0db : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":3},{"id":"3871793967c632f747c5ab6251c99563376e8c7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 17:29:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"775266b6b680d6bcd36f4c0fb4431b74202017b3","revisions":{"3cd459f4e472cd943dfd93155261add8c5242c37":{"kind":"REWORK","_number":1,"created":"2020-03-20 16:05:55.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/60/714160/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/60/714160/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/714160/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/714160/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/714160/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/60/714160/1"}}},"commit":{"parents":[{"commit":"11bd536edc698c9beecebe0d3ca365046024f235","subject":"Rename x/devstack-plugin-nfs to openstack/devstack-plugin-nfs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/11bd536edc698c9beecebe0d3ca365046024f235"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-20 16:05:31.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-20 16:05:31.000000000","tz":-300},"subject":"Stop tracking gerrit upstream","message":"Stop tracking gerrit upstream\n\nWe\u0027re not doing local dev any more and dont\u0027 need this. Plus,\nit\u0027s breaking zuul.\n\nChange-Id: I815c71fc7e924bd7fcfbe7851caad92d370da812\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3cd459f4e472cd943dfd93155261add8c5242c37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3cd459f4e472cd943dfd93155261add8c5242c37"}]},"branch":"refs/heads/master"},"7d8432d83ae1a582b3cfecc5e8ca16fead610572":{"kind":"REWORK","_number":2,"created":"2020-03-20 16:12:04.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/60/714160/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/60/714160/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/714160/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/714160/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/714160/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/60/714160/2"}}},"commit":{"parents":[{"commit":"11bd536edc698c9beecebe0d3ca365046024f235","subject":"Rename x/devstack-plugin-nfs to openstack/devstack-plugin-nfs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/11bd536edc698c9beecebe0d3ca365046024f235"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-20 16:05:31.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-20 16:12:00.000000000","tz":-300},"subject":"Don\u0027t load zuul objects from opendev/gerrit","message":"Don\u0027t load zuul objects from opendev/gerrit\n\nWe added config to gerrit upstream, which is now flowing down\nand confusing our zuul. We don\u0027t have in-repo job config anyway\nsince we\u0027re building images in system-config.\n\nChange-Id: I815c71fc7e924bd7fcfbe7851caad92d370da812\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7d8432d83ae1a582b3cfecc5e8ca16fead610572"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7d8432d83ae1a582b3cfecc5e8ca16fead610572"}]},"branch":"refs/heads/master"},"775266b6b680d6bcd36f4c0fb4431b74202017b3":{"kind":"REWORK","_number":3,"created":"2020-03-20 16:21:08.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/60/714160/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/60/714160/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/714160/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/714160/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/60/714160/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/60/714160/3"}}},"commit":{"parents":[{"commit":"11bd536edc698c9beecebe0d3ca365046024f235","subject":"Rename x/devstack-plugin-nfs to openstack/devstack-plugin-nfs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/11bd536edc698c9beecebe0d3ca365046024f235"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-20 16:05:31.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-20 16:21:04.000000000","tz":-300},"subject":"Don\u0027t load zuul objects from opendev/gerrit","message":"Don\u0027t load zuul objects from opendev/gerrit\n\nWe added config to gerrit upstream, which is now flowing down\nand confusing our zuul. We don\u0027t have in-repo job config anyway\nsince we\u0027re building images in system-config.\n\nChange-Id: I815c71fc7e924bd7fcfbe7851caad92d370da812\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/775266b6b680d6bcd36f4c0fb4431b74202017b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/775266b6b680d6bcd36f4c0fb4431b74202017b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
