)]}'
{"id":"openstack%2Fproject-config~714478","triplet_id":"openstack%2Fproject-config~master~Ide4bdf1de6a8ce5d1f86ea82a9fbf5b8a8bf437d","project":"openstack/project-config","branch":"master","topic":"project-rename","hashtags":[],"change_id":"Ide4bdf1de6a8ce5d1f86ea82a9fbf5b8a8bf437d","subject":"Add whitebox-tempest-plugin under QA","status":"MERGED","created":"2020-03-23 15:41:30.000000000","updated":"2020-06-12 22:41:50.000000000","submitted":"2020-06-12 21:37:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714478-1591997868332-332a2ac9","meta_rev_id":"ed4dbb323c37d30df0889929500e70ade23cef23","_number":714478,"virtual_id_number":714478,"owner":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"date":"2020-06-12 14:56:08.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-12 21:37:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2020-06-11 20:57:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-03-28 16:21:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2020-03-27 14:18:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":1,"date":"2020-06-12 21:30:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-24 13:49:51.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-03-27 14:18:14.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-03-28 16:21:17.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-06-11 20:57:35.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-06-12 14:56:08.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-06-12 21:30:14.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2020-06-12 21:37:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa01f62b8cb95b152d9c31418d421564a3ecc2c2","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-03-23 15:41:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8386806e23359b75713d281c2588b5852d52509","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 15:53:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/de32c0772a8b4c5a9025bbb0d4753976 : SUCCESS in 7m 44s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2261807ceeec473b852919fbe56ac6d2 : FAILURE in 3m 02s","accounts_in_message":[],"_revision_number":1},{"id":"49195faf3688915ff3fa53260e1f787a7fe5bc97","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-03-23 15:59:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7fbf73a7eb85ab408cf11901e96cf6a479bc5704","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-23 16:05:11.000000000","message":"Patch Set 2: Code-Review-1\n\nTo move something from x to openstack, you need to rename it. Follow the process at https://docs.opendev.org/opendev/infra-manual/latest/creators.html#project-renames completely.","accounts_in_message":[],"_revision_number":2},{"id":"d3be89245fdc378b23d5fb131c08539074f9c99b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 16:12:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5b8b779bf426468394809ac1188872e1 : SUCCESS in 7m 34s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8a5d36d2a58d4c63893b73fe389a4557 : FAILURE in 7m 06s","accounts_in_message":[],"_revision_number":2},{"id":"2849cebd933701ae4196b86dd30d22cc985334d1","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-03-23 16:18:51.000000000","message":"Topic set to project-rename","accounts_in_message":[],"_revision_number":2},{"id":"fa007fd5f9f6412d21b3fbd22e1fa8588bde7b3b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-23 16:34:43.000000000","message":"Patch Set 2:\n\nok, this repo is under x namespace. This is a complete example for rename process - https://review.opendev.org/#/c/711834/","accounts_in_message":[],"_revision_number":2},{"id":"4139cf3ad2c608bc5d4d2e59d655d629638478dd","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-03-23 17:06:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0f07732184cd7a7aee5f60e2ee7141f90702d6d1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-23 17:08:31.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nAdd the repo also to gerritbot.\n\ndid you update the wiki?","accounts_in_message":[],"_revision_number":3},{"id":"1f346e5cbfcaee5b378e4609e00eed7fc5aa59f9","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-03-23 17:12:40.000000000","message":"Patch Set 3:\n\n(1 comment)\n\n\u003e (1 comment)\n \u003e \n \u003e Add the repo also to gerritbot.\n\nDone.\n\n \u003e did you update the wiki?\n\nI did: https://wiki.openstack.org/wiki/Meetings/InfraTeamMeeting#Upcoming_Project_Renames","accounts_in_message":[],"_revision_number":3},{"id":"009047da2af216a3a77d8f0dcb5ae74b3192b893","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-03-23 17:12:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3cae8f09173acf23e379e4f521d8b79613f13071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 17:22:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a8e28dbc03ba4b249dcde6a584697927 : SUCCESS in 7m 40s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/9f6d2f45b5124cf0ad5d5df546f4804d : SUCCESS in 5m 25s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/c2aed067e91541848460d61505cda875 : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":4},{"id":"636f41e6cae9f399b2eed620afb23abca1709971","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-23 17:29:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"492897db4addc58e91883bd090e1b05a471b8115","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-03-23 18:41:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d7f622af829b382e52297eb42022900b6a2a6bdf","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-03-24 13:47:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6a5cfaa97a6b6ffe03ceb900967b24ee42e3363b","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-03-24 13:48:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0e88402cbdea7534eb88ad5394c80fc323d701b3","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-03-24 13:49:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ad0ea9b7dd995fd9d5836d458001bb64bb2fff30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 14:00:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/468df3e2967b45aeb25b8429e891c996 : SUCCESS in 8m 00s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/905381da5acc477bbf80c89310d16be8 : SUCCESS in 4m 50s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/88c0b5fee2004e048c734d2bb502f800 : SUCCESS in 8m 56s","accounts_in_message":[],"_revision_number":5},{"id":"1234c866ff82356b78be4d87d18972349f90b917","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-27 14:18:14.000000000","message":"Patch Set 5: Code-Review+1\n\nlgtm.","accounts_in_message":[],"_revision_number":5},{"id":"39cd334834f3d352a1574720cd23889f94a76b2b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-28 16:21:17.000000000","message":"Patch Set 5: Code-Review+2 Workflow-1\n\nLGTM, needs to wait for next scheduled merge downtime, thus WIP.","accounts_in_message":[],"_revision_number":5},{"id":"92e4782732c114279c167112bc2ee67b2140c313","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-19 00:55:47.000000000","message":"Patch Set 5:\n\nany update on this? can we merge it now.","accounts_in_message":[],"_revision_number":5},{"id":"1e857ce3e7e5335a22c515254830b4ad63898fd1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-19 05:17:40.000000000","message":"Patch Set 5:\n\nPlease talk with Clark, this needs a scheduled downtime and will be merged during that downtime.","accounts_in_message":[],"_revision_number":5},{"id":"d22e69afd6763fb58980075245e52827cd0e0f23","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-20 14:26:32.000000000","message":"Patch Set 5:\n\n\u003e Please talk with Clark, this needs a scheduled downtime and will be\n \u003e merged during that downtime.\n\nsure, thanks. I will ping him on IRC.","accounts_in_message":[],"_revision_number":5},{"id":"f3680b32867f8e799db0d8c6667ce25089fe1daa","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-06-11 20:57:35.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"fd95bd7aa51013f93440d4da35f2c43bb127f5f2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-12 08:21:29.000000000","message":"Patch Set 5: -Workflow","accounts_in_message":[],"_revision_number":5},{"id":"8e377cb73c99a2207e0062b23222d0ae78501f83","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-06-12 14:56:08.000000000","message":"Patch Set 5:\n\nscheduled for today: http://lists.opendev.org/pipermail/service-announce/2020-June/000004.html","accounts_in_message":[],"_revision_number":5},{"id":"8ddea18204b9a4e19f4cb39553582db0cd5887cf","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-06-12 14:57:04.000000000","message":"Patch Set 5: Workflow-1\n\nRename is scheduled for later today.","accounts_in_message":[],"_revision_number":5},{"id":"00928a3c604b398d5847e3b8decf7810ca472528","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-06-12 21:30:14.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a5a63f6842a6b715f2ca5121f92c1bc27ed8795c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 21:30:26.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"cf87564d7a9982e0da42876e950d8cf19e0deb72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 21:37:47.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/caf7386aa1584d1ab40b64980e1c57e2 : SUCCESS in 5m 42s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/6a4d1c997c5048ea9c24e2410a51612a : SUCCESS in 3m 30s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/f5606a3ea7694647b16770dd22fac336 : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":5},{"id":"83d530d188278c492cc4c60e93f700a48a48df24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 21:37:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"ed4dbb323c37d30df0889929500e70ade23cef23","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 22:41:50.000000000","message":"Patch Set 5:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/eafe6fc8b7154f57bdd982dd058e9936 : SUCCESS in 2m 10s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/41632670eba2403f9c8ef3b59d8ad5d1 : SUCCESS in 10m 01s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"710f9c824e6c7126431fdca592e52e7eaf8875c9","revisions":{"4d828af98fc366d1d37e5c0a8193c282e24f4a31":{"kind":"REWORK","_number":1,"created":"2020-03-23 15:41:30.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/78/714478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/714478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/714478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/714478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/714478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/714478/1"}}},"commit":{"parents":[{"commit":"7b440dac3a86c71d15d3466a41ff9c4f9d6f3f34","subject":"Merge \"check-release-approval: handle no-review case\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7b440dac3a86c71d15d3466a41ff9c4f9d6f3f34"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-03-23 15:39:28.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-03-23 15:39:28.000000000","tz":-240},"subject":"Add whitebox-tempest-plugin under QA","message":"Add whitebox-tempest-plugin under QA\n\nTempest defines its scope as only what is accessible through the\nvarious REST APIs. Some cloud features cannot be properly tested when\nusing only the REST API. The whitebox-tempest-plugin is a Tempest\nplugin whose scope explicitly requires peeking behind the curtain. In\nother words, if a feature or behavior can be fully tested using only a\nREST API, such a test does not belong in whitebox-tempest-plugin. On\nthe other hand, if fully testing a feature or behavior requires\naccessing the control plane like a human operator or admin would, such\na test belongs in whitebox-tempest plugin.\n\nNow that the spec [1] for whitebox has merged, this patch moves the\nwhitebox-tempest-plugin from x to openstack (QA).\n\n[1] https://review.opendev.org/#/c/656890/\n\nChange-Id: Ide4bdf1de6a8ce5d1f86ea82a9fbf5b8a8bf437d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4d828af98fc366d1d37e5c0a8193c282e24f4a31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4d828af98fc366d1d37e5c0a8193c282e24f4a31"}]},"branch":"refs/heads/master"},"a7ab65661929beac511a6d92b6b6924aca2605f7":{"kind":"REWORK","_number":2,"created":"2020-03-23 15:59:44.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/78/714478/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/714478/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/714478/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/714478/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/714478/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/714478/2"}}},"commit":{"parents":[{"commit":"7b440dac3a86c71d15d3466a41ff9c4f9d6f3f34","subject":"Merge \"check-release-approval: handle no-review case\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7b440dac3a86c71d15d3466a41ff9c4f9d6f3f34"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-03-23 15:39:28.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-03-23 15:59:35.000000000","tz":-240},"subject":"Add whitebox-tempest-plugin under QA","message":"Add whitebox-tempest-plugin under QA\n\nTempest defines its scope as only what is accessible through the\nvarious REST APIs. Some cloud features cannot be properly tested when\nusing only the REST API. The whitebox-tempest-plugin is a Tempest\nplugin whose scope explicitly requires peeking behind the curtain. In\nother words, if a feature or behavior can be fully tested using only a\nREST API, such a test does not belong in whitebox-tempest-plugin. On\nthe other hand, if fully testing a feature or behavior requires\naccessing the control plane like a human operator or admin would, such\na test belongs in whitebox-tempest plugin.\n\nNow that the spec [1] for whitebox has merged, this patch moves the\nwhitebox-tempest-plugin from x to openstack (QA).\n\n[1] https://review.opendev.org/#/c/656890/\n\nChange-Id: Ide4bdf1de6a8ce5d1f86ea82a9fbf5b8a8bf437d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a7ab65661929beac511a6d92b6b6924aca2605f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a7ab65661929beac511a6d92b6b6924aca2605f7"}]},"branch":"refs/heads/master"},"88f62b85d069c450282d68aaf845789f6bf1853d":{"kind":"REWORK","_number":3,"created":"2020-03-23 17:06:08.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/78/714478/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/714478/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/714478/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/714478/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/714478/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/714478/3"}}},"commit":{"parents":[{"commit":"7b440dac3a86c71d15d3466a41ff9c4f9d6f3f34","subject":"Merge \"check-release-approval: handle no-review case\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7b440dac3a86c71d15d3466a41ff9c4f9d6f3f34"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-03-23 15:39:28.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-03-23 17:05:57.000000000","tz":-240},"subject":"Add whitebox-tempest-plugin under QA","message":"Add whitebox-tempest-plugin under QA\n\nTempest defines its scope as only what is accessible through the\nvarious REST APIs. Some cloud features cannot be properly tested when\nusing only the REST API. The whitebox-tempest-plugin is a Tempest\nplugin whose scope explicitly requires peeking behind the curtain. In\nother words, if a feature or behavior can be fully tested using only a\nREST API, such a test does not belong in whitebox-tempest-plugin. On\nthe other hand, if fully testing a feature or behavior requires\naccessing the control plane like a human operator or admin would, such\na test belongs in whitebox-tempest plugin.\n\nNow that the spec [1] for whitebox has merged, this patch moves the\nwhitebox-tempest-plugin from x to openstack (QA).\n\n[1] https://review.opendev.org/#/c/656890/\n\nChange-Id: Ide4bdf1de6a8ce5d1f86ea82a9fbf5b8a8bf437d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/88f62b85d069c450282d68aaf845789f6bf1853d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/88f62b85d069c450282d68aaf845789f6bf1853d"}]},"branch":"refs/heads/master"},"9432e2d6a3eac52e693e7ac7444b162972329f12":{"kind":"REWORK","_number":4,"created":"2020-03-23 17:12:47.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/78/714478/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/714478/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/714478/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/714478/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/714478/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/714478/4"}}},"commit":{"parents":[{"commit":"7b440dac3a86c71d15d3466a41ff9c4f9d6f3f34","subject":"Merge \"check-release-approval: handle no-review case\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7b440dac3a86c71d15d3466a41ff9c4f9d6f3f34"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-03-23 15:39:28.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-03-23 17:12:04.000000000","tz":-240},"subject":"Add whitebox-tempest-plugin under QA","message":"Add whitebox-tempest-plugin under QA\n\nTempest defines its scope as only what is accessible through the\nvarious REST APIs. Some cloud features cannot be properly tested when\nusing only the REST API. The whitebox-tempest-plugin is a Tempest\nplugin whose scope explicitly requires peeking behind the curtain. In\nother words, if a feature or behavior can be fully tested using only a\nREST API, such a test does not belong in whitebox-tempest-plugin. On\nthe other hand, if fully testing a feature or behavior requires\naccessing the control plane like a human operator or admin would, such\na test belongs in whitebox-tempest plugin.\n\nNow that the spec [1] for whitebox has merged, this patch moves the\nwhitebox-tempest-plugin from x to openstack (QA).\n\n[1] https://review.opendev.org/#/c/656890/\n\nChange-Id: Ide4bdf1de6a8ce5d1f86ea82a9fbf5b8a8bf437d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9432e2d6a3eac52e693e7ac7444b162972329f12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9432e2d6a3eac52e693e7ac7444b162972329f12"}]},"branch":"refs/heads/master"},"710f9c824e6c7126431fdca592e52e7eaf8875c9":{"kind":"REWORK","_number":5,"created":"2020-03-24 13:47:53.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/78/714478/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/714478/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/714478/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/714478/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/714478/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/714478/5"}}},"commit":{"parents":[{"commit":"7b440dac3a86c71d15d3466a41ff9c4f9d6f3f34","subject":"Merge \"check-release-approval: handle no-review case\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7b440dac3a86c71d15d3466a41ff9c4f9d6f3f34"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-03-23 15:39:28.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-03-24 13:47:28.000000000","tz":-240},"subject":"Add whitebox-tempest-plugin under QA","message":"Add whitebox-tempest-plugin under QA\n\nTempest defines its scope as only what is accessible through the\nvarious REST APIs. Some cloud features cannot be properly tested when\nusing only the REST API. The whitebox-tempest-plugin is a Tempest\nplugin whose scope explicitly requires peeking behind the curtain. In\nother words, if a feature or behavior can be fully tested using only a\nREST API, such a test does not belong in whitebox-tempest-plugin. On\nthe other hand, if fully testing a feature or behavior requires\naccessing the control plane like a human operator or admin would, such\na test belongs in whitebox-tempest plugin.\n\nNow that the spec [1] for whitebox has merged, this patch moves the\nwhitebox-tempest-plugin from x to openstack (QA).\n\n[1] https://review.opendev.org/#/c/656890/\n\nChange-Id: Ide4bdf1de6a8ce5d1f86ea82a9fbf5b8a8bf437d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/710f9c824e6c7126431fdca592e52e7eaf8875c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/710f9c824e6c7126431fdca592e52e7eaf8875c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
