)]}'
{"id":"openstack%2Fproject-config~714540","triplet_id":"openstack%2Fproject-config~master~Ib9c1ec47024a93571b3c1289ba3ab537417e6830","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Ib9c1ec47024a93571b3c1289ba3ab537417e6830","subject":"Apply gerrit jobs directly to jeepyb","status":"MERGED","created":"2020-03-23 21:53:38.000000000","updated":"2020-03-23 22:56:45.000000000","submitted":"2020-03-23 22:56:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714540-1585004205398-6b47a46b","meta_rev_id":"29903d9b337a385d23bc73043f4440df10babf85","_number":714540,"virtual_id_number":714540,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-23 22:56:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2020-03-23 22:41:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2020-03-23 22:05:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2020-03-23 22:41:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-23 22:05:37.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-03-23 22:41:18.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2020-03-23 22:56:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9df69b0e5799b23c5b73d0f6515c0d088eb47adc","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-23 21:53:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da555fb7670097625a24cf772d34effe4a84715d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 21:54:02.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  expected str for dictionary value @ data[\u0027gate\u0027][\u0027jobs\u0027][3][\u0027system-config-upload-image-gerrit-2.15\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      name: opendev/jeepyb\n      templates:\n        - publish-to-pypi\n      check:\n        jobs:\n          - opendev-buildset-registry\n          - system-config-build-image-gerrit-base:\n              dependencies:\n                - name: opendev-buildset-registry\n                - name: system-config-build-image-python-builder\n                  soft: true\n          - system-config-build-image-gerrit-2.13:\n              dependencies:\n                - name: opendev-buildset-registry\n                - name: system-config-build-image-python-builder\n                  soft: true\n                - name: system-config-build-image-gerrit-base\n                  soft: true\n          - system-config-build-image-gerrit-2.15:\n              dependencies:\n                - name: opendev-buildset-registry\n                - name: system-config-build-image-gerrit-base\n                  soft: true\n          - system-config-build-image-gerrit-2.16:\n              dependencies:\n                - name: opendev-buildset-registry\n                - name: system-config-build-image-gerrit-base\n                  soft: true\n          - system-config-build-image-gerrit-3.0:\n              dependencies:\n                - name: opendev-buildset-registry\n                - name: system-config-build-image-gerrit-base\n                  soft: true\n          - system-config-build-image-gerrit-master:\n              voting: false\n              dependencies:\n                - name: opendev-buildset-registry\n                - name: system-config-build-image-gerrit-base\n                  soft: true\n      gate:\n        jobs:\n          - opendev-buildset-registry\n          - system-config-upload-image-gerrit-base:\n              dependencies:\n                - name: opendev-buildset-registry\n                - name: system-config-upload-image-python-builder\n                  soft: true\n          - system-config-upload-image-gerrit-2.13:\n              dependencies:\n                - name: opendev-buildset-registry\n                - name: system-config-upload-image-python-builder\n                  soft: true\n                - name: system-config-upload-image-gerrit-base\n                  soft: true\n          - system-config-upload-image-gerrit-2.15:\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 201, column 3","accounts_in_message":[],"_revision_number":1},{"id":"af7eaf0ba078ce73aa8f1cea8db3ec1664d562b8","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-23 22:00:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b6fb858c03ebd0e01821f1d3e2775a57f0f0180d","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-03-23 22:05:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2b2eecdb75750c62fad9cdd816f9567682b9bd6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 22:09:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/202f6134baca49858a2d13ebce6bdbc1 : SUCCESS in 7m 55s","accounts_in_message":[],"_revision_number":2},{"id":"4c45250b3661bb62fb766bd7eb168285cf403d9f","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-03-23 22:41:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fdf67e22b0b454a82c271a2b3e4d3df922643d2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 22:41:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9735f0bc6c43554c52346efb023adbcf29318b99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 22:56:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/771c51991c55447eb081e2907ae12415 : SUCCESS in 8m 51s","accounts_in_message":[],"_revision_number":2},{"id":"29903d9b337a385d23bc73043f4440df10babf85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 22:56:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ba3645eb084efbb4f39677f498d3fd23e57191b8","revisions":{"8b7d2032f654118e9c48a42e57e19c63c93c963c":{"kind":"REWORK","_number":1,"created":"2020-03-23 21:53:38.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/40/714540/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/40/714540/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/714540/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/714540/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/714540/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/40/714540/1"}}},"commit":{"parents":[{"commit":"bb51a1affc63cbb6ce7c77dfaec8be9d664ca2bf","subject":"Run gerrit image jobs on jeepyb patches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bb51a1affc63cbb6ce7c77dfaec8be9d664ca2bf"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-23 21:52:22.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-23 21:52:22.000000000","tz":-300},"subject":"Apply gerrit jobs directly to jeepyb","message":"Apply gerrit jobs directly to jeepyb\n\nThere is currently a zuul bug that is attaching runnability to\nproject-template definition and so it\u0027s seeing that since this\nproject-template is defined in system-config which is not a config\nproject that jeepyb can\u0027t run the upload jobs. That\u0027s not what\nwe want to happen.\n\nFor now, work around it by just expanding the template here. Once\nthe zuul bug is fixed, we can revert this change.\n\nChange-Id: Ib9c1ec47024a93571b3c1289ba3ab537417e6830\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8b7d2032f654118e9c48a42e57e19c63c93c963c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8b7d2032f654118e9c48a42e57e19c63c93c963c"}]},"branch":"refs/heads/master"},"ba3645eb084efbb4f39677f498d3fd23e57191b8":{"kind":"REWORK","_number":2,"created":"2020-03-23 22:00:32.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/40/714540/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/40/714540/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/714540/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/714540/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/714540/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/40/714540/2"}}},"commit":{"parents":[{"commit":"bb51a1affc63cbb6ce7c77dfaec8be9d664ca2bf","subject":"Run gerrit image jobs on jeepyb patches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bb51a1affc63cbb6ce7c77dfaec8be9d664ca2bf"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-23 21:52:22.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-03-23 22:00:26.000000000","tz":-300},"subject":"Apply gerrit jobs directly to jeepyb","message":"Apply gerrit jobs directly to jeepyb\n\nThere is currently a zuul bug that is attaching runnability to\nproject-template definition and so it\u0027s seeing that since this\nproject-template is defined in system-config which is not a config\nproject that jeepyb can\u0027t run the upload jobs. That\u0027s not what\nwe want to happen.\n\nFor now, work around it by just expanding the template here. Once\nthe zuul bug is fixed, we can revert this change.\n\nChange-Id: Ib9c1ec47024a93571b3c1289ba3ab537417e6830\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ba3645eb084efbb4f39677f498d3fd23e57191b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ba3645eb084efbb4f39677f498d3fd23e57191b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
