)]}'
{"id":"openstack%2Fproject-config~715733","triplet_id":"openstack%2Fproject-config~master~Ia35b793486acc0d21155b6f1a4a48137c64fd262","project":"openstack/project-config","branch":"master","topic":"pep8-inline-comments","hashtags":[],"change_id":"Ia35b793486acc0d21155b6f1a4a48137c64fd262","subject":"DNM: making sure pep8 inline comments don\u0027t break","status":"ABANDONED","created":"2020-03-29 21:52:56.000000000","updated":"2020-03-30 12:23:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d08f0a930e3457bb55331670741b4d61a0865bb5","_number":715733,"virtual_id_number":715733,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-03-30 06:01:47.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-03-29 22:00:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-29 22:00:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-30 06:01:47.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"9b0af1671818b4eb4bf50cf7a96f367e8c5991a9","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-03-29 21:52:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14ce741fef769b9982b37e000616d89dd4435714","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 22:00:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a2f0d1b7ad0840069465c6598a0d9376 : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":1},{"id":"ee541ec96729b8c050c1905f3c8a7415dd7632ab","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-30 06:01:47.000000000","message":"Patch Set 1:\n\nDNM here will never work since this is a config-project.\n\nBetter use openstack-zuul-jobs.","accounts_in_message":[],"_revision_number":1},{"id":"468a1450c7665af6eda9de486fa851d8ef32c7c6","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-03-30 12:23:58.000000000","message":"Abandoned\n\ngood call","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5d94c28621041a417f2743c28223fda072c5745d","revisions":{"5d94c28621041a417f2743c28223fda072c5745d":{"kind":"REWORK","_number":1,"created":"2020-03-29 21:52:56.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/33/715733/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/33/715733/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/715733/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/715733/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/715733/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/33/715733/1"}}},"commit":{"parents":[{"commit":"cc567d89e9e6f17dc56b903cc6c978b5e687d075","subject":"Always build gerrit images on jeepyb changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cc567d89e9e6f17dc56b903cc6c978b5e687d075"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-03-29 21:51:48.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-03-29 21:51:48.000000000","tz":-240},"subject":"DNM: making sure pep8 inline comments don\u0027t break","message":"DNM: making sure pep8 inline comments don\u0027t break\n\nDepends-On: https://review.opendev.org/715727\nChange-Id: Ia35b793486acc0d21155b6f1a4a48137c64fd262\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5d94c28621041a417f2743c28223fda072c5745d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5d94c28621041a417f2743c28223fda072c5745d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
