)]}'
{"id":"openstack%2Fproject-config~715734","triplet_id":"openstack%2Fproject-config~master~I418e4581b764ff1577b0eeda25ff11220c3703d8","project":"openstack/project-config","branch":"master","topic":"new-project","hashtags":[],"change_id":"I418e4581b764ff1577b0eeda25ff11220c3703d8","subject":"Add new project and repository for tripleo-compute-extras","status":"ABANDONED","created":"2020-03-29 21:53:16.000000000","updated":"2020-05-22 10:26:35.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"11c8584987b6e34316af769d8f09d42991767345","_number":715734,"virtual_id_number":715734,"owner":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"date":"2020-04-20 15:42:05.000000000","_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-03-31 10:17:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-03-31 11:14:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-29 21:55:52.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-03-29 21:56:03.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-29 21:56:22.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2020-03-30 10:38:43.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-03-30 12:53:02.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-03-30 16:50:17.000000000","updated_by":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"reviewer":{"_account_id":10379,"name":"Rajini Karthik","email":"rrajini@gmail.com","username":"rajinir"},"state":"REVIEWER"},{"updated":"2020-03-31 10:17:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-31 11:14:26.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"state":"REVIEWER"},{"updated":"2020-04-20 15:42:05.000000000","updated_by":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"reviewer":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"state":"REVIEWER"}],"messages":[{"id":"f19dd5536e7e9414d0533f4b60a52776b2a171dd","author":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"date":"2020-03-29 21:53:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"913edfd11570abbc4bd67747c22f332575e08699","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 22:00:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/31ba195ee9ae41cdb6e0cf55a16383b4 : SUCCESS in 6m 19s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0d7edbb056e34ba187b94d805b803be2 : FAILURE in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"c79215873a483aad6e27f5c1e86bf99699590d2b","author":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"date":"2020-03-29 22:12:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"35be6fa24bdcffd7df6dabc35d731c1cb60501a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 22:24:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/eb920fc05ffb4736bd5c5d067baf41e7 : SUCCESS in 7m 06s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7d8991cf821e4c6b8e7ed6aed1b66cb1 : FAILURE in 4m 43s","accounts_in_message":[],"_revision_number":2},{"id":"63f8f2a5d37f08df92dc7a191fa57759d2613348","author":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"date":"2020-03-29 22:27:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5708136ee420e5e3c1955166883142914fe9ff40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 22:35:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9f9cd0efda304ffa9320e1e53f83e580 : SUCCESS in 6m 30s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c91d3a8eec4945ca922516e3d5120142 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":3},{"id":"a29b2d3b3e48ff105ffb7477b387c4c78fddd7a5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-30 05:42:36.000000000","message":"Patch Set 3:\n\nwhy is this in the x namespace and not in openstack? Shouldn\u0027t it be part of TripleO?","accounts_in_message":[],"_revision_number":3},{"id":"6d21c5279427c7dbd00f30aa8d66378d24a1df0d","author":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"date":"2020-03-30 09:07:08.000000000","message":"Patch Set 3:\n\n\u003e why is this in the x namespace and not in openstack? Shouldn\u0027t it\n \u003e be part of TripleO?\n\nNo, I\u0027ve consulted that with TripleO cores and since this will include code that handle 3rd party kind of stuff they did not want this to be part of TripleO. Similar to https://review.opendev.org/#/c/711114/","accounts_in_message":[],"_revision_number":3},{"id":"6e132a006d7c671331a10392be8c378d7d7af616","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-03-30 10:33:07.000000000","message":"Patch Set 3:\n\nThe \"x\" namespace was originally meant to hold repositories without a clear ownership... I\u0027m not sure what the current OpenDev team stance is on it, but IMHO we should not create new repositories in \u0027x\u0027, and favor creating new namespaces.","accounts_in_message":[],"_revision_number":3},{"id":"8262c34a24b6bc8945fec967d8561526c579c499","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-30 10:38:43.000000000","message":"Patch Set 3:\n\nWe accept for x - but it\u0027s really a last resort... Creation of new namespaces is fine.","accounts_in_message":[],"_revision_number":3},{"id":"49c62666370d800259f05815deb2d15a693c5f53","author":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"date":"2020-03-30 11:18:07.000000000","message":"Patch Set 3:\n\n\u003e We accept for x - but it\u0027s really a last resort... Creation of new\n \u003e namespaces is fine.\n\nIMHO it would be overkill to create a new namespace for hosting few Ansible roles related to Nova features alone. Since the content of the repo will support features related to 3rd party hardware we could not include those into `tripleo-ansible` repo and since this is deployment related it does not really fit into Nova repo. I\u0027ve picked \u0027x\u0027 namespace because I think it does not have clear ownership.\n\nHowever, I can create separate namespace for this if that\u0027s what you advice.","accounts_in_message":[],"_revision_number":3},{"id":"fb98e1e48df128374f43a4f0c686eab1d8aabc50","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-03-30 12:53:02.000000000","message":"Patch Set 3: Code-Review-1\n\ni think this should be a deliverable of the ooo project and i don\u0027t think FPGA support or NVDIMM support should be consider third party.\n\nvNVDIMMs for guest is now a supported feature of nova and FPGA support is part of cyborg which we shoudl finish integrating with nova this cycle i think both should be enabled through roles here https://github.com/openstack/tripleo-ansible/tree/master/tripleo_ansible/roles","accounts_in_message":[],"_revision_number":3},{"id":"87ad2cee3e52bc81c5de9182f654bbf625db483a","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2020-03-30 13:43:21.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\nSuggest adding some data what info will TripleO needed for deploying and LCM of OpenStack to use.","accounts_in_message":[],"_revision_number":3},{"id":"66d27798748f93445de6888ec76dacc8d617b185","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-03-30 18:09:39.000000000","message":"Topic changed from create-tripleo-compute-3rdparty to new-project","accounts_in_message":[],"_revision_number":3},{"id":"1857d6e596ead08351556cf015970ef03846bc7e","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2020-03-31 10:08:57.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"b87877d182bd208ee64e49c2449296541a652945","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 10:17:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/69cf994f707f4b2dba8932a8abb91749 : SUCCESS in 6m 34s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2d688e85488d49bb8d4037742a27db0a : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":4},{"id":"9f58eb29f310e168c1e703913ff97c1a46f73648","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2020-03-31 11:14:26.000000000","message":"Patch Set 4: Code-Review-1","accounts_in_message":[],"_revision_number":4},{"id":"e0db8994cf7df25c4fe05ad35a6e1e1d6abf2d56","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2020-03-31 11:15:23.000000000","message":"Patch Set 4:\n\nWIP.","accounts_in_message":[],"_revision_number":4},{"id":"07996210164aba71af40535226a0e53d1d2441b2","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2020-04-20 15:42:05.000000000","message":"Patch Set 4:\n\nso what is the plan for these extras?","accounts_in_message":[],"_revision_number":4},{"id":"11c8584987b6e34316af769d8f09d42991767345","author":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"date":"2020-05-22 10:26:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c91a6bf52161706480c9550f30a080a440c5ec0c","revisions":{"f21299cb4293b2708b7ee3392b2674f843960460":{"kind":"REWORK","_number":1,"created":"2020-03-29 21:53:16.000000000","uploader":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"ref":"refs/changes/34/715734/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/34/715734/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/715734/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/715734/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/715734/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/34/715734/1"}}},"commit":{"parents":[{"commit":"192cdd5e746590d07aacc596f455261231066b26","subject":"Merge \"Added new project openstack-tempest-skiplist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/192cdd5e746590d07aacc596f455261231066b26"}]}],"author":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-03-29 21:40:21.000000000","tz":120},"committer":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-03-29 21:52:32.000000000","tz":120},"subject":"Add new project and repository for tripleo-compute-extras","message":"Add new project and repository for tripleo-compute-extras\n\nThis commit creates a new project and repository\n`tripleo-compute-extras` which will contain Ansible roles\nused to support advanced Nova features (e.g FPGA, NVDIMM) in TripleO.\n\nChange-Id: I418e4581b764ff1577b0eeda25ff11220c3703d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f21299cb4293b2708b7ee3392b2674f843960460"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f21299cb4293b2708b7ee3392b2674f843960460"}]},"branch":"refs/heads/master"},"bee6ec56e1dcfe504e04d2d7ca64891920275641":{"kind":"REWORK","_number":2,"created":"2020-03-29 22:12:15.000000000","uploader":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"ref":"refs/changes/34/715734/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/34/715734/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/715734/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/715734/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/715734/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/34/715734/2"}}},"commit":{"parents":[{"commit":"192cdd5e746590d07aacc596f455261231066b26","subject":"Merge \"Added new project openstack-tempest-skiplist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/192cdd5e746590d07aacc596f455261231066b26"}]}],"author":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-03-29 21:40:21.000000000","tz":120},"committer":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-03-29 22:11:51.000000000","tz":120},"subject":"Add new project and repository for tripleo-compute-extras","message":"Add new project and repository for tripleo-compute-extras\n\nThis commit creates a new project and repository\n`tripleo-compute-extras` which will contain Ansible roles\nused to support advanced Nova features (e.g FPGA, NVDIMM) in TripleO.\n\nChange-Id: I418e4581b764ff1577b0eeda25ff11220c3703d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bee6ec56e1dcfe504e04d2d7ca64891920275641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bee6ec56e1dcfe504e04d2d7ca64891920275641"}]},"branch":"refs/heads/master"},"96bb5f65568a8248753e77510806ddd21877276b":{"kind":"REWORK","_number":3,"created":"2020-03-29 22:27:54.000000000","uploader":{"_account_id":30133,"name":"Piotr Kopec","email":"pkopec17@gmail.com","username":"pkopec"},"ref":"refs/changes/34/715734/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/34/715734/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/715734/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/715734/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/715734/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/34/715734/3"}}},"commit":{"parents":[{"commit":"192cdd5e746590d07aacc596f455261231066b26","subject":"Merge \"Added new project openstack-tempest-skiplist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/192cdd5e746590d07aacc596f455261231066b26"}]}],"author":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-03-29 21:40:21.000000000","tz":120},"committer":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-03-29 22:27:37.000000000","tz":120},"subject":"Add new project and repository for tripleo-compute-extras","message":"Add new project and repository for tripleo-compute-extras\n\nThis commit creates a new project and repository\n`tripleo-compute-extras` which will contain Ansible roles\nused to support advanced Nova features (e.g FPGA, NVDIMM) in TripleO.\n\nChange-Id: I418e4581b764ff1577b0eeda25ff11220c3703d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/96bb5f65568a8248753e77510806ddd21877276b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/96bb5f65568a8248753e77510806ddd21877276b"}]},"branch":"refs/heads/master"},"c91a6bf52161706480c9550f30a080a440c5ec0c":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-03-31 10:08:57.000000000","uploader":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"ref":"refs/changes/34/715734/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/34/715734/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/715734/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/715734/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/715734/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/34/715734/4"}}},"commit":{"parents":[{"commit":"192cdd5e746590d07aacc596f455261231066b26","subject":"Merge \"Added new project openstack-tempest-skiplist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/192cdd5e746590d07aacc596f455261231066b26"}]}],"author":{"name":"Piotr Kopec","email":"pkopec@redhat.com","date":"2020-03-29 21:40:21.000000000","tz":120},"committer":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2020-03-31 10:08:51.000000000","tz":330},"subject":"Add new project and repository for tripleo-compute-extras","message":"Add new project and repository for tripleo-compute-extras\n\nThis commit creates a new project and repository\n`tripleo-compute-extras` which will contain Ansible roles\nused to support advanced Nova features like FPGA Remote\nsystem update in TripleO.\n\nChange-Id: I418e4581b764ff1577b0eeda25ff11220c3703d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c91a6bf52161706480c9550f30a080a440c5ec0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c91a6bf52161706480c9550f30a080a440c5ec0c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
