)]}'
{"id":"openstack%2Fproject-config~716159","triplet_id":"openstack%2Fproject-config~master~I222a148a30dc155f4a6fc442f9678c575aa79a67","project":"openstack/project-config","branch":"master","topic":"project-yaml-normalization","hashtags":[],"change_id":"I222a148a30dc155f4a6fc442f9678c575aa79a67","subject":"Normalize projects.yaml","status":"MERGED","created":"2020-03-31 06:08:12.000000000","updated":"2020-04-01 14:04:48.000000000","submitted":"2020-04-01 14:01:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716159-1585749704576-d25266a9","meta_rev_id":"30a864ad601b4575723ffc6dca575dc7760127fc","_number":716159,"virtual_id_number":716159,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-01 14:01:44.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-04-01 13:45:56.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-04-01 13:50:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2020-04-01 13:50:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-01 13:45:56.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-04-01 13:50:46.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-01 14:01:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"177384f6307aef478eb9458d9b12400995014d8f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-03-31 06:08:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1e73689c203cc7af48a3b1af2c4d44a91ccf4e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 06:28:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9119b7104306492e809d9fdf780bac69 : SUCCESS in 9m 39s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7c6fe6496ebc4f28b6b54c24ea879b6e : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":1},{"id":"90ef4c61a76d8903e53960ac3ae037c1a033a31c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-31 06:58:39.000000000","message":"Patch Set 1: Workflow-1\n\nrepos are not correctly imported","accounts_in_message":[],"_revision_number":1},{"id":"bc6a64a95e78ba53e07c8fe3b7a21ee15206b9b2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-04-01 06:03:40.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"681af3a35f6119d6761879e304b63825ccd51528","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 06:31:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5ecbc32b25094b9c8eb575596f538993 : SUCCESS in 10m 48s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5adf7f5fbfef4ee0be993eb25d100c6f : SUCCESS in 7m 30s","accounts_in_message":[],"_revision_number":2},{"id":"bf98ab5895fbb86fe0caa86cb6a5367f1e217edf","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-04-01 13:45:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a5ef548e9bc20a7d0b1fcb825bfd8c466807aa29","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-01 13:50:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nrepos imported","accounts_in_message":[],"_revision_number":2},{"id":"8c3d232a9b41d3e17025f81ea5d3cf7679790740","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 13:50:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"414829cba5a2c4332893beb81c804372fb5a0e15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 14:01:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"927a3c4039b66ed80579bda3dd53bcdb6897af29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 14:01:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/28e258f640774ae0bf1b9d635ae053f7 : SUCCESS in 8m 48s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3424d10a2dfa45feb8a0d0fc87839bfa : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":2},{"id":"67de342289b19233a4b1977b28e8d5047154eea1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 14:04:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/a37619b0281d48e3880ef23586ba1fa6 : SUCCESS in 2m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"37ee543582bbb17efe12d312e713a17aa75867e5","revisions":{"00181e95f11cc5c09972eefb21fb20facd5d65a5":{"kind":"REWORK","_number":1,"created":"2020-03-31 06:08:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/59/716159/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/716159/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/716159/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/716159/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/716159/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/716159/1"}}},"commit":{"parents":[{"commit":"b2799c12a25b7afb18d73afdb8334d142f5b07a5","subject":"Merge \"Add Cert-Manager Armada app to StarlingX\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b2799c12a25b7afb18d73afdb8334d142f5b07a5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-31 06:08:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-03-31 06:08:09.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: I222a148a30dc155f4a6fc442f9678c575aa79a67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/00181e95f11cc5c09972eefb21fb20facd5d65a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/00181e95f11cc5c09972eefb21fb20facd5d65a5"}]},"branch":"refs/heads/master"},"37ee543582bbb17efe12d312e713a17aa75867e5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-01 06:03:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/59/716159/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/716159/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/716159/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/716159/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/716159/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/716159/2"}}},"commit":{"parents":[{"commit":"a311ab927484edb0195475b306fff9c18941bb7b","subject":"Remove upstream from xstatic-dagre","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a311ab927484edb0195475b306fff9c18941bb7b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-01 06:03:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-04-01 06:03:38.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: I222a148a30dc155f4a6fc442f9678c575aa79a67\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/37ee543582bbb17efe12d312e713a17aa75867e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/37ee543582bbb17efe12d312e713a17aa75867e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
