)]}'
{"id":"openstack%2Fproject-config~716785","triplet_id":"openstack%2Fproject-config~master~Ie1ca9bda749e17cdd5269850c03feca100395f1f","project":"openstack/project-config","branch":"master","topic":"fedora-31-patch1","hashtags":[],"change_id":"Ie1ca9bda749e17cdd5269850c03feca100395f1f","subject":"zuul-worker: remove python-apt \u0026 libselinux deps","status":"MERGED","created":"2020-04-01 22:58:47.000000000","updated":"2020-04-03 16:03:46.000000000","submitted":"2020-04-03 16:03:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716785-1585929812292-d5d81bf2","meta_rev_id":"fea4b92e0cd4bc2c30a3b7c3d8f159d2e5f4c426","_number":716785,"virtual_id_number":716785,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-03 16:03:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-04-03 15:24:40.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-04-03 15:51:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2020-04-03 15:51:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-03 15:24:40.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-04-03 15:51:30.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-03 16:03:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5a0327b74535ddc2e90f92cf07f9c28a1df5bbaa","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-04-01 22:58:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54e3cedac3787d7121a35b844be34fb0629ad814","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-04-01 23:01:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9f5d59f495dd72d9fee5ef91018e3547d29fb4d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 23:13:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/576325b5f8494ffcb1b459fc39e97b5a : SUCCESS in 7m 23s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/986e34e904334a54b0b8e976dc005e8e : FAILURE in 3m 22s","accounts_in_message":[],"_revision_number":2},{"id":"5c6a01b47f7141b9c27a32e59a027277fa7dc11d","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-04-03 14:10:48.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"892a6aaeaea272460a6640aa3f13b4c0dad944ed","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-04-03 14:20:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"33c2e9b44a0d4550824c7936bf6bdbc2917c4732","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 15:01:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a9e40d8f04494d63af5d6387c734de26 : SUCCESS in 8m 59s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/ac4465a3f3ea45359e151387c7f41f47 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":3},{"id":"98c53ed5089d248a5dddd8beda0f960336297075","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-04-03 15:24:40.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a174510b7e5b77ace666a03108bba08ffc109cd8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-03 15:51:30.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2d14685a490bae5c9ff8bded98d1b6c2ae1b376d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 15:51:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8aff5303ac23fb434bc796d97dc7907e17a356cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 16:03:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6d9ba73e837e449d9e6151384a330be9 : SUCCESS in 10m 28s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/0ad954af57ee4b54aea933c3fb2cebce : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":3},{"id":"e16190df2a5a3634d46822cea4695195f1326fe5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 16:03:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"2ea33c6a0f607af9eedbd99405c3a3514de70080","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 16:03:46.000000000","message":"Patch Set 3:\n\nUnable to freeze job graph: Job infra-prod-service-nodepool depends on infra-prod-update-system-config which was not run.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8a1b382301f047cc6c5ce7b02023184a5a4374be","revisions":{"bec903a82743fdb3df2e39abcb9823ee380919da":{"kind":"REWORK","_number":1,"created":"2020-04-01 22:58:47.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/85/716785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/85/716785/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/85/716785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/85/716785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/85/716785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/85/716785/1"}}},"commit":{"parents":[{"commit":"539b1a3b6c73e5779c40a731616b55359165b414","subject":"zuul-worker element: use python3-libselinux for fedora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/539b1a3b6c73e5779c40a731616b55359165b414"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-01 22:45:45.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-01 22:45:45.000000000","tz":660},"subject":"zuul-worker: remove python-apt \u0026 libselinux deps","message":"zuul-worker: remove python-apt \u0026 libselinux deps\n\nThese have been here since the initial commit\nI0e76931fdb4ca0c7445b1e72dc348f0cf03eaee5 in 2016.\n\nThese have now bifurcated into python3 packages and are brought in\nelsewhere; we don\u0027t need to additionally bring it in here.  At best we\nduplicate the install, at worse we get it wrong\n(I563896f1caa1a68781f10b27cd7d9cf2606aaa17).\n\nlibselinux is installed for all platforms by the redhat-common package\n[1].  This element probably pre-dates that.\n\npython[3]-apt is brought in by software-properties-common [2] for all\nplatforms, which will bring in the correct python[3] version.\n\n[1] https://opendev.org/openstack/diskimage-builder/src/branch/master/diskimage_builder/elements/redhat-common/package-installs.yaml#L14\n[2] https://opendev.org/openstack/diskimage-builder/src/branch/master/diskimage_builder/elements/base/pre-install.d/03-baseline-tools\n\nChange-Id: Ie1ca9bda749e17cdd5269850c03feca100395f1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bec903a82743fdb3df2e39abcb9823ee380919da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bec903a82743fdb3df2e39abcb9823ee380919da"}]},"branch":"refs/heads/master"},"823d05e9fc1c6f0ecec43d5c6551a4f12ecbe22a":{"kind":"REWORK","_number":2,"created":"2020-04-01 23:01:36.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/85/716785/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/85/716785/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/85/716785/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/85/716785/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/85/716785/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/85/716785/2"}}},"commit":{"parents":[{"commit":"539b1a3b6c73e5779c40a731616b55359165b414","subject":"zuul-worker element: use python3-libselinux for fedora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/539b1a3b6c73e5779c40a731616b55359165b414"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-01 22:45:45.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-01 23:01:24.000000000","tz":660},"subject":"zuul-worker: remove python-apt \u0026 libselinux deps","message":"zuul-worker: remove python-apt \u0026 libselinux deps\n\nThese have been here since the initial commit\nI0e76931fdb4ca0c7445b1e72dc348f0cf03eaee5 in 2016.\n\nThese have now bifurcated into python3 packages and are brought in\nelsewhere; we don\u0027t need to additionally bring it in here.  At best we\nduplicate the install, at worse we get it wrong\n(I563896f1caa1a68781f10b27cd7d9cf2606aaa17).\n\nlibselinux is installed for all platforms by the redhat-common package\n[1].  This element probably pre-dates that.\n\npython[3]-apt is brought in by software-properties-common [2] for all\nplatforms, which will bring in the correct python[3] version.\n\n[1] https://opendev.org/openstack/diskimage-builder/src/branch/master/diskimage_builder/elements/redhat-common/package-installs.yaml#L14\n[2] https://opendev.org/openstack/diskimage-builder/src/branch/master/diskimage_builder/elements/base/pre-install.d/03-baseline-tools\n\nChange-Id: Ie1ca9bda749e17cdd5269850c03feca100395f1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/823d05e9fc1c6f0ecec43d5c6551a4f12ecbe22a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/823d05e9fc1c6f0ecec43d5c6551a4f12ecbe22a"}]},"branch":"refs/heads/master"},"8a1b382301f047cc6c5ce7b02023184a5a4374be":{"kind":"REWORK","_number":3,"created":"2020-04-03 14:20:26.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/85/716785/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/85/716785/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/85/716785/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/85/716785/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/85/716785/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/85/716785/3"}}},"commit":{"parents":[{"commit":"539b1a3b6c73e5779c40a731616b55359165b414","subject":"zuul-worker element: use python3-libselinux for fedora","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/539b1a3b6c73e5779c40a731616b55359165b414"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-01 22:45:45.000000000","tz":660},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-04-03 14:20:05.000000000","tz":-300},"subject":"zuul-worker: remove python-apt \u0026 libselinux deps","message":"zuul-worker: remove python-apt \u0026 libselinux deps\n\nThese have been here since the initial commit\nI0e76931fdb4ca0c7445b1e72dc348f0cf03eaee5 in 2016.\n\nThese have now bifurcated into python3 packages and are brought in\nelsewhere; we don\u0027t need to additionally bring it in here.  At best we\nduplicate the install, at worse we get it wrong\n(I563896f1caa1a68781f10b27cd7d9cf2606aaa17).\n\nlibselinux is installed for all platforms by the redhat-common package\n[1].  This element probably pre-dates that.\n\npython[3]-apt is brought in by software-properties-common [2] for all\nplatforms, which will bring in the correct python[3] version.\n\n[1] https://opendev.org/openstack/diskimage-builder/src/branch/master/diskimage_builder/elements/redhat-common/package-installs.yaml#L14\n[2] https://opendev.org/openstack/diskimage-builder/src/branch/master/diskimage_builder/elements/base/pre-install.d/03-baseline-tools\n\nChange-Id: Ie1ca9bda749e17cdd5269850c03feca100395f1f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8a1b382301f047cc6c5ce7b02023184a5a4374be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8a1b382301f047cc6c5ce7b02023184a5a4374be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
