)]}'
{"id":"openstack%2Fproject-config~717004","triplet_id":"openstack%2Fproject-config~master~I050464cc4ad896342ffc9125d12af73058791438","project":"openstack/project-config","branch":"master","topic":"infra-prod-zuul","hashtags":[],"change_id":"I050464cc4ad896342ffc9125d12af73058791438","subject":"Run service-nameserver playbook on zone file changes","status":"MERGED","created":"2020-04-02 13:35:38.000000000","updated":"2020-04-06 14:37:57.000000000","submitted":"2020-04-06 14:37:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717004-1586183877749-73754ddb","meta_rev_id":"b93d1be58b17cbd13a7ea76d1e52560b3f9e7ec5","_number":717004,"virtual_id_number":717004,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-06 14:37:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-04-02 22:31:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-04-03 08:48:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2020-04-06 14:24:08.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 20:48:13.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-04-03 08:48:08.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-06 14:37:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"827a8b0756f6aa62ba6c7ef0f43834716e8f49f1","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-04-02 13:35:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a63a8dc45975a4859ee141d2e0e75d3f90266f54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 14:06:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3b7a18a7f1da4192a678f9cf1625eb7e : SUCCESS in 10m 29s","accounts_in_message":[],"_revision_number":1},{"id":"5989a81b17e450dce8f5f299adee2974323d1724","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-04-02 20:48:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4e0e9bd1df963a4d3651472245a67dc8b153c9b6","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-04-02 20:54:24.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"93ca43ebb0fafc8f3c1c9ea8449c5db175cced48","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-04-02 20:56:40.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"9e0075f3f52262ab08d501758de6d83f95450a84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 21:35:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/eecd51a60a794b57ae2049b06e1991dc : SUCCESS in 10m 25s","accounts_in_message":[],"_revision_number":3},{"id":"936346fb9fb3978a89b59872d68812c0e942b75a","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-04-02 22:31:14.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"0295507ded2408ca3251e3fe60ac39ba1adf3fb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 22:52:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/57c7e4e4bc63473f9f4c7a7011b99ef4 : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":4},{"id":"615e9a278b64f48dd67beffb8d8928a98020dbbe","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-03 08:48:08.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM, not approving since dependency is not merged. Please self-approve when ready.","accounts_in_message":[],"_revision_number":4},{"id":"722077c3be923c75a3203ac15b6c8ea64b8da7e9","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-04-06 14:24:08.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b260dd023fd2fa8a07f374ab17119ed46f92eda3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 14:24:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d69b42dcf9c3ffa6e64b7c7b119d78862e179c38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 14:37:57.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5ad70a2a8903485eb3c96156ad5ef717 : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":4},{"id":"ed5f304191d3b4a3fc0b5c27b1775ff36553a2c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 14:37:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c60521b840b92c8267dde85034bfcddcff2b1658","revisions":{"0bd858cfe73de7433b850b0d8860dfd697080b0e":{"kind":"REWORK","_number":1,"created":"2020-04-02 13:35:38.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/04/717004/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/04/717004/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/717004/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/717004/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/717004/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/04/717004/1"}}},"commit":{"parents":[{"commit":"167b161b7ef769aa387c6f89a81644557a491f1b","subject":"Add Fedora 31 to main config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/167b161b7ef769aa387c6f89a81644557a491f1b"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-04-02 13:32:09.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-04-02 13:35:12.000000000","tz":-300},"subject":"Run service-nameserver playbook on zone file changes","message":"Run service-nameserver playbook on zone file changes\n\nWe\u0027ve got the keys in place, and the zone file repos are, for now,\nowned by infra-root, so this should be safe.\n\nDepends-On: https://review.opendev.org/716764\nChange-Id: I050464cc4ad896342ffc9125d12af73058791438\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0bd858cfe73de7433b850b0d8860dfd697080b0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0bd858cfe73de7433b850b0d8860dfd697080b0e"}]},"branch":"refs/heads/master"},"f6db224afd2c03653a2097b43041ddf3741dc7e5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-02 20:54:24.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/04/717004/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/04/717004/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/717004/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/717004/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/717004/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/04/717004/2"}}},"commit":{"parents":[{"commit":"341396142f39479fa8fdf6923da992a2f39798ae","subject":"Be clear that zone-zuul-ci.org is owned by infra-core","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/341396142f39479fa8fdf6923da992a2f39798ae"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-04-02 13:32:09.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-04-02 20:54:15.000000000","tz":-300},"subject":"Run service-nameserver playbook on zone file changes","message":"Run service-nameserver playbook on zone file changes\n\nWe\u0027ve got the keys in place, and the zone file repos are, for now,\nowned by infra-root, so this should be safe.\n\nDepends-On: https://review.opendev.org/716764\nChange-Id: I050464cc4ad896342ffc9125d12af73058791438\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f6db224afd2c03653a2097b43041ddf3741dc7e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f6db224afd2c03653a2097b43041ddf3741dc7e5"}]},"branch":"refs/heads/master"},"fcf6c4e593b14f8b129884057ec92c098fa8707f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-02 20:56:40.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/04/717004/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/04/717004/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/717004/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/717004/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/717004/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/04/717004/3"}}},"commit":{"parents":[{"commit":"5038a9191fcfb9de1940820dca6cfeb90506adcd","subject":"Be clear that zone repos are owned by infra-core","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5038a9191fcfb9de1940820dca6cfeb90506adcd"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-04-02 13:32:09.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-04-02 20:56:36.000000000","tz":-300},"subject":"Run service-nameserver playbook on zone file changes","message":"Run service-nameserver playbook on zone file changes\n\nWe\u0027ve got the keys in place, and the zone file repos are, for now,\nowned by infra-root, so this should be safe.\n\nDepends-On: https://review.opendev.org/716764\nChange-Id: I050464cc4ad896342ffc9125d12af73058791438\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fcf6c4e593b14f8b129884057ec92c098fa8707f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fcf6c4e593b14f8b129884057ec92c098fa8707f"}]},"branch":"refs/heads/master"},"c60521b840b92c8267dde85034bfcddcff2b1658":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-04-02 22:31:14.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/04/717004/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/04/717004/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/717004/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/717004/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/717004/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/04/717004/4"}}},"commit":{"parents":[{"commit":"5e49aacb60f417f1b0b1a8e0f61e029fad6fe425","subject":"Trigger infra-prod-service-nodepool on nodepool changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5e49aacb60f417f1b0b1a8e0f61e029fad6fe425"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-04-02 13:32:09.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-04-02 22:30:33.000000000","tz":-300},"subject":"Run service-nameserver playbook on zone file changes","message":"Run service-nameserver playbook on zone file changes\n\nWe\u0027ve got the keys in place, and the zone file repos are, for now,\nowned by infra-root, so this should be safe.\n\nDepends-On: https://review.opendev.org/716764\nChange-Id: I050464cc4ad896342ffc9125d12af73058791438\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c60521b840b92c8267dde85034bfcddcff2b1658"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c60521b840b92c8267dde85034bfcddcff2b1658"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
