)]}'
{"id":"openstack%2Fproject-config~717330","triplet_id":"openstack%2Fproject-config~master~I301750a91039311675fbb0c5fa2a76349f547b62","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I301750a91039311675fbb0c5fa2a76349f547b62","subject":"Trigger nodepool run","status":"MERGED","created":"2020-04-03 16:26:57.000000000","updated":"2020-04-03 16:59:10.000000000","submitted":"2020-04-03 16:58:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717330-1585933135603-40a9e122","meta_rev_id":"65e5e684679762c401b31cfdecb4d3dfef80c7dd","_number":717330,"virtual_id_number":717330,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-03 16:58:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-03 16:43:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-03 16:43:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-03 16:43:57.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-04-03 16:58:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc81489b1f392fc309db62ab2d7a3c8e47196f1e","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-04-03 16:26:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b8b1792625b61a34c24a60f13fea4782cedee44","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-04-03 16:43:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nEmpty commit, these are not the droids you\u0027re looking for.","accounts_in_message":[],"_revision_number":1},{"id":"d478439674b552beacc046b6a1e2913ec4ce13ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 16:44:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/b6559a65aea84dbf8c471115ed5def0d : SUCCESS in 5m 20s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e05e4dd8405146c7b353b671fbd6db97 : SUCCESS in 7m 49s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/fca104b13a204ad4a47e4188441ec5d0 : SUCCESS in 4m 08s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f2de47ef414c4fd4be223c2913f8950d : SUCCESS in 5m 27s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/0a3bd0a177154429a6b36d9f11c20491 : SUCCESS in 4m 02s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/13ba4af1b58249cfb09212465c907b20 : SUCCESS in 4m 13s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/1ef226c5bc6a4fa5ba3f1c2b3c710fa8 : SUCCESS in 8m 50s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/3974545cd86447e5bc41213ab4318d53 : SUCCESS in 6m 22s\n- project-config-bindep-fallback-centos-7 https://zuul.opendev.org/t/openstack/build/1ad9a1e9f62547c18769c80b9b1f56bc : SUCCESS in 3m 17s\n- project-config-bindep-fallback-opensuse-15 https://zuul.opendev.org/t/openstack/build/e065d5af063f4399ac6d36a582829922 : SUCCESS in 5m 26s\n- project-config-bindep-fallback-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/851c469493064a3aa3e14c3d6e12d763 : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":1},{"id":"8aac6c1c3637736cbd51524afc27253b315f74bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 16:44:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"efc15b5c8bfcee75b09523d8855d1fcbb1fde046","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 16:58:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/3acaa654ca724c5ab746a3abe5fbfd5d : SUCCESS in 4m 19s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/da96c491f4b246c7ad894666714d6b56 : SUCCESS in 8m 48s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/62bf90087f67442eb1f1a4e00ac5a005 : SUCCESS in 6m 04s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/bdcd133530614a2d996d54044137406a : SUCCESS in 4m 17s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/2456b60cb892463a935e10bc6f45280c : SUCCESS in 3m 22s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/32844bd49673423eb07a573e1f560243 : SUCCESS in 4m 01s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/43729708b6394093a480a47d2d539fb1 : SUCCESS in 7m 52s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/ae1d068c70a34bc0a90a51ec5b35e3a3 : SUCCESS in 4m 52s\n- project-config-bindep-fallback-centos-7 https://zuul.opendev.org/t/openstack/build/708bc24b09084c2b9fe38edaab41d1d4 : SUCCESS in 3m 44s\n- project-config-bindep-fallback-opensuse-15 https://zuul.opendev.org/t/openstack/build/e4af6020d3744c8db760b4c2e0b2779f : SUCCESS in 4m 32s\n- project-config-bindep-fallback-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/2b8041f6260a4ee4a3b2ae70d04159e5 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"68b84e1a20a9f69528802cbc2744c36881b7b908","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 16:58:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"65e5e684679762c401b31cfdecb4d3dfef80c7dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 16:59:10.000000000","message":"Patch Set 1:\n\nUnable to freeze job graph: Job infra-prod-letsencrypt not defined","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2db1c7768425c790079e5c3247b7039f5f2f868","revisions":{"b2db1c7768425c790079e5c3247b7039f5f2f868":{"kind":"REWORK","_number":1,"created":"2020-04-03 16:26:57.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/30/717330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/717330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/717330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/717330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/717330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/717330/1"}}},"commit":{"parents":[{"commit":"5f199f5bbe5adb55fb6b589df81e89e15c0512f2","subject":"Add update-system-config to promote list for project-config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5f199f5bbe5adb55fb6b589df81e89e15c0512f2"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-04-03 16:26:14.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-04-03 16:26:14.000000000","tz":-300},"subject":"Trigger nodepool run","message":"Trigger nodepool run\n\nThe previous patch fixed the promote job for running nodepool\nansible. This should trigger it to run, which was skipped in the\nprevious patch.\n\nChange-Id: I301750a91039311675fbb0c5fa2a76349f547b62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b2db1c7768425c790079e5c3247b7039f5f2f868"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b2db1c7768425c790079e5c3247b7039f5f2f868"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
