)]}'
{"id":"openstack%2Fproject-config~717715","triplet_id":"openstack%2Fproject-config~master~I3901963eab6845e302feb3d3bca5ee47697a5658","project":"openstack/project-config","branch":"master","topic":"osa/retire_pip_build","hashtags":[],"change_id":"I3901963eab6845e302feb3d3bca5ee47697a5658","subject":"Retire repo_build and pip_install roles","status":"MERGED","created":"2020-04-06 11:27:46.000000000","updated":"2020-04-07 08:07:06.000000000","submitted":"2020-04-07 08:07:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717715-1586246826412-dc9227fa","meta_rev_id":"04a6ee6e5d69eee31b1c5ede513976624b9f1cc9","_number":717715,"virtual_id_number":717715,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"date":"2020-04-06 14:59:50.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-07 08:07:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-04-07 04:54:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-04-07 07:58:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-04-07 07:58:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 04:54:46.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-07 07:58:29.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-04-07 08:07:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c70356ec7b23fcc959fbcce8df2d3d5bbbdcc991","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-06 11:27:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c13769669677473125855786a785e91793a1374","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 11:37:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4ed0c0cf885a4784bf465327f06c5407 : SUCCESS in 7m 38s","accounts_in_message":[],"_revision_number":1},{"id":"13f49cc3a717d85ecefae910938629fa0d741e11","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-06 12:06:25.000000000","message":"Patch Set 1:\n\nCould you push all changes together, please? I\u0027d like to review together to ensure nothing is missed.\n\nThe change itself is fine, will +2 once I saw everything.","accounts_in_message":[],"_revision_number":1},{"id":"6e7b09f3fd0c4f8b6b1834027aa79b22e8924e88","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-06 14:48:49.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"899545b904f696ec707ba7418abfc6bdd6d312e8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-06 14:59:50.000000000","message":"Patch Set 2:\n\nAdded as a follow-up patch","accounts_in_message":[],"_revision_number":2},{"id":"41d1d1d5f0bd9b92a477ce5e9068e302d67b0df0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 15:00:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2e121aec2dd340f3b6341942e77748b8 : SUCCESS in 8m 59s","accounts_in_message":[],"_revision_number":2},{"id":"370e92f46d97cefec54c129f8e6bb5e5edee7104","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-07 04:54:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ebdd0ed0e96eec45a6da21209386fc9a8e40cbc8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-04-07 07:58:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nSeems weird having to add this stuff in order to retire it, but that\u0027s the way it is I guess.","accounts_in_message":[],"_revision_number":2},{"id":"038516ea8a6140e7d39ea18294f3133be3fec7bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 07:58:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"44b39830594ff0f7d9d999d02036a1bd91a0f595","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 08:07:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4fa3f2ac1d2c3dcedeb303e6a58d96f9a7e5a025","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 08:07:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6e40f2ba34f24f3a8a44e03bac30f93c : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b37825720a27e9fbb91fd150574cbf9af49f3f64","revisions":{"2aaaaaba98d4b8a24d9756b7d2743fc8e95029e3":{"kind":"REWORK","_number":1,"created":"2020-04-06 11:27:46.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/15/717715/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/15/717715/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/717715/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/717715/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/717715/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/15/717715/1"}}},"commit":{"parents":[{"commit":"a8a41f162f469489f670a552cf3d42b521eceae8","subject":"Install gpg tooling on dib images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a8a41f162f469489f670a552cf3d42b521eceae8"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-06 11:27:35.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-06 11:27:35.000000000","tz":180},"subject":"Retire repo_build and pip_install roles","message":"Retire repo_build and pip_install roles\n\nAs both repo_build and pip_install roles have been succesfully replaced\nwith python_venv_build role, we have no need in it\u0027s futher maintaining\nand releasing them.\n\nChange-Id: I3901963eab6845e302feb3d3bca5ee47697a5658\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2aaaaaba98d4b8a24d9756b7d2743fc8e95029e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2aaaaaba98d4b8a24d9756b7d2743fc8e95029e3"}]},"branch":"refs/heads/master"},"b37825720a27e9fbb91fd150574cbf9af49f3f64":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-04-06 14:48:49.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/15/717715/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/15/717715/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/717715/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/717715/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/15/717715/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/15/717715/2"}}},"commit":{"parents":[{"commit":"a8a41f162f469489f670a552cf3d42b521eceae8","subject":"Install gpg tooling on dib images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a8a41f162f469489f670a552cf3d42b521eceae8"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-06 11:27:35.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-04-06 14:48:45.000000000","tz":0},"subject":"Retire repo_build and pip_install roles","message":"Retire repo_build and pip_install roles\n\nAs both repo_build and pip_install roles have been succesfully replaced\nwith python_venv_build role, we have no need in it\u0027s futher maintaining\nand releasing them.\n\nNeeded-By: https://review.opendev.org/717717\nNeeded-By: https://review.opendev.org/716389\nChange-Id: I3901963eab6845e302feb3d3bca5ee47697a5658\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b37825720a27e9fbb91fd150574cbf9af49f3f64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b37825720a27e9fbb91fd150574cbf9af49f3f64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
