)]}'
{"id":"openstack%2Fproject-config~718479","triplet_id":"openstack%2Fproject-config~master~Ie7b4f520d6f47d56a71c812dcc06fd5d26da8fe0","project":"openstack/project-config","branch":"master","topic":"per-project-mirroring","hashtags":[],"change_id":"Ie7b4f520d6f47d56a71c812dcc06fd5d26da8fe0","subject":"Introduce job for granular GitHub mirroring","status":"MERGED","created":"2020-04-08 14:48:13.000000000","updated":"2020-04-10 13:30:58.000000000","submitted":"2020-04-09 14:47:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718479-1586443660853-a4cc011a","meta_rev_id":"a9841823ffc85d423290deff297a8fcc1e5c35e6","_number":718479,"virtual_id_number":718479,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-09 14:47:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-04-09 13:46:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-09 14:36:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-09 14:36:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-09 13:46:43.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-09 14:36:33.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-04-09 14:47:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"00e449e297d6b47203eb5423bf28eb1652c8cdbf","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-08 14:48:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"690b2697e0c6dcb3f015cba756d09923f6327d07","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-08 14:52:26.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nLGTM, want to use a repo that merges more changes than reno?","accounts_in_message":[],"_revision_number":1},{"id":"d51292d4b1153ba158356ab201880c5dbdd720f0","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-04-08 15:08:18.000000000","message":"Patch Set 1: Code-Review+2\n\nAndreas Jaeger wrote:\n\u003e LGTM, want to use a repo that merges more changes than reno?\n\nMaybe openstack/release-test would allow more flexibility as a test sandbox, since what you\u0027re really going to want to test out actively are corner cases like branch creation/deletion and tag propagation.","accounts_in_message":[],"_revision_number":1},{"id":"a6ee519d7cece6af658a344dfe37bcf01a8e8dfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 15:13:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/815ef9747a684054ac8af3431edbcfcc : SUCCESS in 9m 48s","accounts_in_message":[],"_revision_number":1},{"id":"d3d260d9e3c457671e6a128e46abaee46289aa74","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-09 13:45:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1ac73395e73cd2e12f38d2c37fe20c2f14f4d70e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-09 13:46:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b7251f471488e220a4e0110b5f02ac660404416e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 13:55:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/01c6b3fe8a394bc59623686d39e65f40 : SUCCESS in 8m 19s","accounts_in_message":[],"_revision_number":2},{"id":"2a463ee070eed2e59c9e107bd26f27613dd3fa3b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-04-09 14:36:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cbd138d06c685b28779a1916678660cdd4581321","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 14:36:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2ddc20660fe1cbc8e7fe68adfc593dd9973934d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 14:47:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a3221e8591eb5a1a6e984dc071fbffdf86e839c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 14:47:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8ed45ed2c6914e6a87fc79ece10a4ed8 : SUCCESS in 9m 15s","accounts_in_message":[],"_revision_number":2},{"id":"0487fad2448cf6e3c9210b7bb0191174e27d6547","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-04-10 00:03:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b954254d6ea05211c3fe0c1a2432e44db9b30c58","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-10 07:13:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7ad2eb7d37e349627d6f8d8ca43b3fabfd5f1b27","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-10 08:46:00.000000000","message":"Patch Set 2:\n\nAh ok... My understanding after asking around was that the job (as defined) would only be available through openstack/project-config, so x/nova could be blocked there by openstack-infra reviewers... I may have understood wrong though.","accounts_in_message":[],"_revision_number":2},{"id":"954c853480906b1739f939d2567ed703cd75c7ca","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-04-10 13:30:58.000000000","message":"Patch Set 2:\n\nClark had suggested in IRC that it should be marked protected, I simply forgot that when reviewing your change. It needs to be added to projects in the config repo for other reasons though, related to implicit branch matchers and a desire to have it running on tag events for branched repositories.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"341dccfb5d9279a9f87ac05450c34dc2368b4132","revisions":{"ab53f02568f3235a09b60d1efd86c4534b7246cc":{"kind":"REWORK","_number":1,"created":"2020-04-08 14:48:13.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/79/718479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/79/718479/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/718479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/718479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/718479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/79/718479/1"}}},"commit":{"parents":[{"commit":"91c02e031d1aa3d595bf010839c10042e05e0029","subject":"Merge \"Retire repo_build and pip_install roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/91c02e031d1aa3d595bf010839c10042e05e0029"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-07 13:57:54.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-08 14:43:41.000000000","tz":120},"subject":"Introduce job for granular GitHub mirroring","message":"Introduce job for granular GitHub mirroring\n\nCreate a openstack-mirror-on-github job that uses the\nopenstack-mirroring user on GitHub (and its SSH key) to\nindividually mirror repositories to the \"openstack\"\norganization on GitHub.\n\nThat will allow us to turn off global replication of the\n\u0027openstack/* repositories at Gerrit-level.\n\nEnable the new job on openstack/reno as a test. It should be\nable to run concurrently with Gerrit-wide mirroring as this\nis an indempotent operation.\n\nOnce this merges and is validated, we can apply the job to\nall official/active openstack projects, before disabling the\nGerrit-wide replication (see https://review.opendev.org/718478).\n\nChange-Id: Ie7b4f520d6f47d56a71c812dcc06fd5d26da8fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ab53f02568f3235a09b60d1efd86c4534b7246cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ab53f02568f3235a09b60d1efd86c4534b7246cc"}]},"branch":"refs/heads/master"},"341dccfb5d9279a9f87ac05450c34dc2368b4132":{"kind":"REWORK","_number":2,"created":"2020-04-09 13:45:46.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/79/718479/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/79/718479/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/718479/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/718479/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/718479/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/79/718479/2"}}},"commit":{"parents":[{"commit":"91c02e031d1aa3d595bf010839c10042e05e0029","subject":"Merge \"Retire repo_build and pip_install roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/91c02e031d1aa3d595bf010839c10042e05e0029"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-07 13:57:54.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-09 13:30:11.000000000","tz":120},"subject":"Introduce job for granular GitHub mirroring","message":"Introduce job for granular GitHub mirroring\n\nCreate a openstack-mirror-on-github job that uses the\nopenstack-mirroring user on GitHub (and its SSH key) to\nindividually mirror repositories to the \"openstack\"\norganization on GitHub.\n\nThat will allow us to turn off global replication of the\n\u0027openstack/* repositories at Gerrit-level.\n\nEnable the new job on openstack/release-test as a test. It\nshould be able to run concurrently with Gerrit-wide\nmirroring as this is an indempotent operation.\n\nOnce this merges and is validated, we can apply the job to\nall official/active openstack projects, before disabling the\nGerrit-wide replication (see https://review.opendev.org/718478).\n\nChange-Id: Ie7b4f520d6f47d56a71c812dcc06fd5d26da8fe0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/341dccfb5d9279a9f87ac05450c34dc2368b4132"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/341dccfb5d9279a9f87ac05450c34dc2368b4132"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
