)]}'
{"id":"openstack%2Fproject-config~718877","triplet_id":"openstack%2Fproject-config~master~I16bc248a1a370ab4819c8068e05bf6d15a3dd16b","project":"openstack/project-config","branch":"master","topic":"fix-new-project","hashtags":[],"change_id":"I16bc248a1a370ab4819c8068e05bf6d15a3dd16b","subject":"Drop dependencies from manage-projects","status":"MERGED","created":"2020-04-10 03:31:08.000000000","updated":"2020-04-10 16:06:21.000000000","submitted":"2020-04-10 16:06:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718877-1586534781095-7fd7f7c7","meta_rev_id":"5138c6d15af849ffd25981d5130a0562ccb244fc","_number":718877,"virtual_id_number":718877,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-10 16:06:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2020-04-10 15:47:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2020-04-10 15:48:13.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":1,"date":"2020-04-10 15:48:13.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-10 15:47:38.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2020-04-10 15:48:13.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-04-10 16:06:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"af7df528b95e0d018acdaeb1c963a61312b5d48a","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-04-10 03:31:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59e2f2cf4ac72db2ebe7a1776ec77b88386914ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 03:39:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/816460c5a12d4121a0ebff31f14357f5 : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":1},{"id":"b026405db1072f4761e7090b7319176e49a006a4","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-04-10 15:44:30.000000000","message":"Patch Set 1:\n\nThe error is:\n\nUnable to freeze job graph: Job infra-prod-remote-puppet-else depends on infra-prod-update-system-config which was not run.","accounts_in_message":[],"_revision_number":1},{"id":"f4a479ccf0e1b97a3fcc94b9f4f832f39f11be0c","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-04-10 15:46:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"96f0b9aa486fbec40418c587210d642d95996c89","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-04-10 15:47:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"65a328a538d369de73146508955cb5e9c9d269d5","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-04-10 15:48:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c0768528de3589cb55fbd74f871c6dae77f23e53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 15:59:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/72931019a6744cbba6c16400bba96d64 : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":2},{"id":"f84f1b4ad019f0903c3a7754d1cf9426d833a2d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 15:59:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"079f08dc01bc239fa3d3098f0ca07e1ca1edc463","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 16:06:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0ffc3a0b87f2491c9eedf43dabb0d68b : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":2},{"id":"ed2d46e5391d21b37be7ee9212b54fcba432e81b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 16:06:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f2459f87161ea711079035679733b98d026f8368","revisions":{"3f1fd3293af0c9205d79a6be0bcf5565fdeee8d8":{"kind":"REWORK","_number":1,"created":"2020-04-10 03:31:08.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/77/718877/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/77/718877/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/718877/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/718877/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/718877/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/77/718877/1"}}},"commit":{"parents":[{"commit":"04133b4bccb931b403a22ddcd49737e02907dd2a","subject":"Revert \"Introduce job for granular GitHub mirroring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/04133b4bccb931b403a22ddcd49737e02907dd2a"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-04-10 03:29:18.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-04-10 03:29:18.000000000","tz":-240},"subject":"Add infra-prod-update-system-config to deploy","message":"Add infra-prod-update-system-config to deploy\n\nThe infra-prod-update-system-config job is a hard dependency to\nrun the infra-prod-remote-puppet-else job which we don\u0027t run inside\nthe post pipeline, this means that the job fails to run when we make\nchanges to zuul/main.yaml\n\nThis patch adds it so that it runs always in post, ideally, it should\nrun when specific files change _but_ that\u0027s an optimization that can\nbe left to someone who understands this better than I do. :)\n\nChange-Id: I16bc248a1a370ab4819c8068e05bf6d15a3dd16b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3f1fd3293af0c9205d79a6be0bcf5565fdeee8d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3f1fd3293af0c9205d79a6be0bcf5565fdeee8d8"}]},"branch":"refs/heads/master"},"f2459f87161ea711079035679733b98d026f8368":{"kind":"REWORK","_number":2,"created":"2020-04-10 15:46:51.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/77/718877/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/77/718877/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/718877/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/718877/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/718877/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/77/718877/2"}}},"commit":{"parents":[{"commit":"04133b4bccb931b403a22ddcd49737e02907dd2a","subject":"Revert \"Introduce job for granular GitHub mirroring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/04133b4bccb931b403a22ddcd49737e02907dd2a"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-04-10 03:29:18.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-04-10 15:45:45.000000000","tz":-240},"subject":"Drop dependencies from manage-projects","message":"Drop dependencies from manage-projects\n\nThe parent job for manage-projects requires a dependency of updating\nthe system-config code which is not necessary when it runs in the\ncontext of openstack/project-config.\n\nChange-Id: I16bc248a1a370ab4819c8068e05bf6d15a3dd16b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f2459f87161ea711079035679733b98d026f8368"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f2459f87161ea711079035679733b98d026f8368"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
