)]}'
{"id":"openstack%2Fproject-config~720316","triplet_id":"openstack%2Fproject-config~master~I84dae0990cd7e90e7d23caf0c3bdfb6cd8abb138","project":"openstack/project-config","branch":"master","topic":"diskimages-globals","hashtags":[],"change_id":"I84dae0990cd7e90e7d23caf0c3bdfb6cd8abb138","subject":"Add ubuntu-bionic-plain to all regions","status":"MERGED","created":"2020-04-16 00:48:40.000000000","updated":"2020-04-17 05:52:40.000000000","submitted":"2020-04-17 05:47:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720316-1587102472154-b16e9fe7","meta_rev_id":"54cacb72719d283c498228863b3d226eff5595f5","_number":720316,"virtual_id_number":720316,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-17 05:47:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-04-16 23:05:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-04-17 05:40:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2020-04-17 05:40:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-16 10:46:58.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-04-16 23:05:39.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-04-17 05:40:49.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-17 05:47:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ffbfad8695e376853cc98e628fb5052f5e2f4163","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-04-16 00:48:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfbe65834d645af1a5ff586c11291af04cb7be8b","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-04-16 00:52:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"48cfa8140382fd706b1066f9eb94b4dd18438adf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 01:02:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fe303b5f61cf48e99b11edf53beeded5 : SUCCESS in 7m 03s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/24dd54644666405d95a4636adb6222be : FAILURE in 4m 08s","accounts_in_message":[],"_revision_number":2},{"id":"4092e870b9b5ad4ea4e158e097b55e7a082b73a2","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-04-16 02:30:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6f91aa55e82cb18b1b96b2ed67cc07af66765042","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 02:40:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a61d65e6c9cb4b51943c13eacfd5b692 : SUCCESS in 6m 55s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/9aa478491a2646e48d3dea7a516c12ce : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":3},{"id":"53bcbbe720ef2f3cc50473bf5dfa3a73c773f402","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-16 05:42:34.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bf73a6eb2daca482f98fce187c18d2161c4045f6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-16 06:37:26.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"ee8c5393b74d2ca14645b6f6a909b558e968501d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 07:08:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/83e460783b14472b82987dfa03e65db3 : SUCCESS in 8m 09s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/bab5c159dc974840912d1fc94125452f : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":4},{"id":"cc1c68ccb9eb1ef1d970445921d9c1973a61aca0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-04-16 10:46:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ce70f9b5225c635593a7d10852563e0ad592e406","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-16 10:54:47.000000000","message":"Patch Set 4: -Code-Review","accounts_in_message":[],"_revision_number":4},{"id":"1d96f21426e96495eaf444d1eff7f30e69bb6f5f","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-04-16 22:18:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4fb4523fa4b31c4abceb61b7f3a15c778eb6de2b","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-04-16 22:42:01.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"f5fcc7fde8439ab8c4de4781f10c0058fd4267fc","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-04-16 23:05:39.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"4119df01cd8d090ea0f94d3e116115e52e535d3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 23:40:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0dba9a668d294be9a8377eb83c3875d7 : SUCCESS in 7m 20s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/2efe7ba27ea54f5f9a29590d40addbeb : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":5},{"id":"32efbd107bd969541ccae42f43d560b60270612b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-17 05:40:49.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"34f021b703fab6cb6ba5c1e76f2d8d666b14f0c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 05:41:01.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b6e32354a105edddb203c669aea7f2f0633e61b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 05:47:51.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/eaaeb34bb2ac444db1bda4c30dafdec1 : SUCCESS in 6m 05s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/1ced2cd2b0e844b2856b07334f23f5a8 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":5},{"id":"7567b24691d3c907eff162eaf8a999f953ea7345","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 05:47:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"4497ba7490e5c77d96102945cbdf6cc0c30a9f8c","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 05:52:40.000000000","message":"Patch Set 5:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/f932819067ba403b982bec5b4c74036f : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"90d6de37acabd647d24c966620b81466a94b9535","revisions":{"4524e675a2c2d2da4cb40b9c79ee5297d17a776b":{"kind":"REWORK","_number":1,"created":"2020-04-16 00:48:40.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/16/720316/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/720316/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/720316/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/720316/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/720316/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/720316/1"}}},"commit":{"parents":[{"commit":"a2696c371f10495bfd8060142dd9d18c6668570c","subject":"nodepool: use job inheritance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a2696c371f10495bfd8060142dd9d18c6668570c"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-16 00:47:57.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-16 00:47:57.000000000","tz":600},"subject":"Add ubuntu-bionic-plain to more regions","message":"Add ubuntu-bionic-plain to more regions\n\nNo reason they can\u0027t run this image\n\nChange-Id: I84dae0990cd7e90e7d23caf0c3bdfb6cd8abb138\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4524e675a2c2d2da4cb40b9c79ee5297d17a776b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4524e675a2c2d2da4cb40b9c79ee5297d17a776b"}]},"branch":"refs/heads/master"},"2c7264a6d7140c139db8cb3289081fded3ede6ee":{"kind":"REWORK","_number":2,"created":"2020-04-16 00:52:04.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/16/720316/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/720316/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/720316/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/720316/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/720316/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/720316/2"}}},"commit":{"parents":[{"commit":"a2696c371f10495bfd8060142dd9d18c6668570c","subject":"nodepool: use job inheritance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a2696c371f10495bfd8060142dd9d18c6668570c"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-16 00:47:57.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-16 00:51:36.000000000","tz":600},"subject":"Add ubuntu-bionic-plain to all regions","message":"Add ubuntu-bionic-plain to all regions\n\nWe\u0027ve been testing with this for a while now, so no reason this can\u0027t\nbe deployed more widely.\n\nChange-Id: I84dae0990cd7e90e7d23caf0c3bdfb6cd8abb138\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2c7264a6d7140c139db8cb3289081fded3ede6ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2c7264a6d7140c139db8cb3289081fded3ede6ee"}]},"branch":"refs/heads/master"},"ab337c0e62ec10651a322b14696c4cfdd1e24fb1":{"kind":"REWORK","_number":3,"created":"2020-04-16 02:30:59.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/16/720316/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/720316/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/720316/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/720316/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/720316/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/720316/3"}}},"commit":{"parents":[{"commit":"a2696c371f10495bfd8060142dd9d18c6668570c","subject":"nodepool: use job inheritance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a2696c371f10495bfd8060142dd9d18c6668570c"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-16 00:47:57.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-16 01:09:31.000000000","tz":600},"subject":"Add ubuntu-bionic-plain to all regions","message":"Add ubuntu-bionic-plain to all regions\n\nWe\u0027ve been testing with this for a while now, so no reason this can\u0027t\nbe deployed more widely.\n\nChange-Id: I84dae0990cd7e90e7d23caf0c3bdfb6cd8abb138\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ab337c0e62ec10651a322b14696c4cfdd1e24fb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ab337c0e62ec10651a322b14696c4cfdd1e24fb1"}]},"branch":"refs/heads/master"},"782815a9f5c00373923d6f789884a553f5c6e79b":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-04-16 06:37:26.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/16/720316/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/720316/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/720316/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/720316/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/720316/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/720316/4"}}},"commit":{"parents":[{"commit":"72aae8309ddfb3769fe6a756e6d2b5545257ca3a","subject":"nodepool: use job inheritance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/72aae8309ddfb3769fe6a756e6d2b5545257ca3a"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-16 00:47:57.000000000","tz":600},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-04-16 06:37:26.000000000","tz":0},"subject":"Add ubuntu-bionic-plain to all regions","message":"Add ubuntu-bionic-plain to all regions\n\nWe\u0027ve been testing with this for a while now, so no reason this can\u0027t\nbe deployed more widely.\n\nChange-Id: I84dae0990cd7e90e7d23caf0c3bdfb6cd8abb138\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/782815a9f5c00373923d6f789884a553f5c6e79b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/782815a9f5c00373923d6f789884a553f5c6e79b"}]},"branch":"refs/heads/master"},"90d6de37acabd647d24c966620b81466a94b9535":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-04-16 22:42:01.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/16/720316/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/720316/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/720316/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/720316/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/720316/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/720316/5"}}},"commit":{"parents":[{"commit":"a2bcee7db7114b297504a3c6c9cf4dcdc4bc89a1","subject":"Merge \"nodepool: use job inheritance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a2bcee7db7114b297504a3c6c9cf4dcdc4bc89a1"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-16 00:47:57.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-16 22:41:31.000000000","tz":600},"subject":"Add ubuntu-bionic-plain to all regions","message":"Add ubuntu-bionic-plain to all regions\n\nWe\u0027ve been testing with this for a while now, so no reason this can\u0027t\nbe deployed more widely.\n\nChange-Id: I84dae0990cd7e90e7d23caf0c3bdfb6cd8abb138\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/90d6de37acabd647d24c966620b81466a94b9535"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/90d6de37acabd647d24c966620b81466a94b9535"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
