)]}'
{"id":"openstack%2Fproject-config~720318","triplet_id":"openstack%2Fproject-config~master~Ic29a25527e259a61ca025325b74acbe64a71774c","project":"openstack/project-config","branch":"master","topic":"diskimages-globals","hashtags":[],"change_id":"Ic29a25527e259a61ca025325b74acbe64a71774c","subject":"nodepool: Add more plain images","status":"MERGED","created":"2020-04-16 01:07:28.000000000","updated":"2020-04-17 08:32:50.000000000","submitted":"2020-04-17 08:25:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720318-1587111902107-cc994f1b","meta_rev_id":"2581758c1c00535902641b7b5d1a12e0a018d6e1","_number":720318,"virtual_id_number":720318,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-17 08:25:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-04-16 23:10:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-04-16 22:42:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-04-17 08:06:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-04-17 08:06:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-16 06:38:10.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-16 23:10:16.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-04-17 08:06:50.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-04-17 08:25:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d1f7437acce50441fe1898fcad792f78ebfe9f60","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-04-16 01:07:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db5dda2ce1f39adfaecdb5cee32c8cf73d7e963a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 01:20:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6da7e1cd33244083b12d2cb7fbdcec05 : SUCCESS in 7m 21s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/4a3b6e0977d54a56ad01e53de44f4c05 : FAILURE in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"520bea5d5fd31dcaaad4ea6edc85c842688a5d65","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-04-16 02:30:59.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"a413e9699f27cde3c0730ee9c128e4352da9e6eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 02:40:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8835a9f417d84199a7ce5a7976ee9c59 : SUCCESS in 6m 32s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/90163d87433140a2933a1d775afde160 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":2},{"id":"f673d3beb125c91367b119c1c55f74d9c0a00792","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-16 06:37:31.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"65173776843c6cc6726bac5e19ab1a1c9382d077","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-16 06:38:10.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6df6d0c6133d7e96e7b3a0c3bebfd9a0b7925013","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 07:21:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/94870f7b35434d6dafc6a8c672799a56 : SUCCESS in 7m 28s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/0a9be86537db417fbd6de41f136bae37 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":3},{"id":"3e59cbf425c713236b4b7dbb5c9ad61d9a3eb86e","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-04-16 22:42:01.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"6c4211ef2cfafbf697f6866d5c5ba4418d5e0055","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-04-16 23:10:16.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ea743485a269f39b61d92f15896267077811c5a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 23:33:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1912d41ccbdc4360b85f719b03e4f7d7 : SUCCESS in 6m 52s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/839d8d48469d44fb8db59287a372b35d : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":4},{"id":"e7881026a04338e0c09d6f43e080ec29eb5efc7e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-04-17 08:06:50.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"3a90372aaf2e1081e0f4eafafd665f85cb94a20f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 08:07:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"14b88ef2427e6e7460ac6028f7a900a206764c2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 08:25:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/279b1715265d4b90a1c98ae7507763e8 : SUCCESS in 9m 44s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/dd4ec2bd842a4b5d873ca7065df6b728 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":4},{"id":"32915fa0e34927852fc5070719377f6e1ed402b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 08:25:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"58568b2e1d0a8465db6220b3904f07223dd4c1e8","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 08:32:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/58d39d0c5a3e4b37a290bec0dfd6a84a : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a3f20a13471f1182ee98647e0e2fdd6a6dba5759","revisions":{"6a957cd4971d64a6ad9bdc3f2370a975b990dea0":{"kind":"REWORK","_number":1,"created":"2020-04-16 01:07:28.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/18/720318/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/18/720318/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/720318/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/720318/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/720318/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/18/720318/1"}}},"commit":{"parents":[{"commit":"2c7264a6d7140c139db8cb3289081fded3ede6ee","subject":"Add ubuntu-bionic-plain to all regions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2c7264a6d7140c139db8cb3289081fded3ede6ee"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-16 01:05:03.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-16 01:06:36.000000000","tz":600},"subject":"nodepool: Add more plain images","message":"nodepool: Add more plain images\n\nAdd more images without pip-and-virtualenv we can test on.  Build all\n-plain images on nb04 (the containerised builder) as an additional\ntest of that environment (the existing bionic-plain build is moved\nhere too).\n\nChange-Id: Ic29a25527e259a61ca025325b74acbe64a71774c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6a957cd4971d64a6ad9bdc3f2370a975b990dea0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6a957cd4971d64a6ad9bdc3f2370a975b990dea0"}]},"branch":"refs/heads/master"},"fb69a2208fd425e13b3683529287b994ec100ce2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-16 02:30:59.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/18/720318/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/18/720318/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/720318/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/720318/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/720318/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/18/720318/2"}}},"commit":{"parents":[{"commit":"ab337c0e62ec10651a322b14696c4cfdd1e24fb1","subject":"Add ubuntu-bionic-plain to all regions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ab337c0e62ec10651a322b14696c4cfdd1e24fb1"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-16 01:05:03.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-16 01:09:33.000000000","tz":600},"subject":"nodepool: Add more plain images","message":"nodepool: Add more plain images\n\nAdd more images without pip-and-virtualenv we can test on.  Build all\n-plain images on nb04 (the containerised builder) as an additional\ntest of that environment (the existing bionic-plain build is moved\nhere too).\n\nChange-Id: Ic29a25527e259a61ca025325b74acbe64a71774c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fb69a2208fd425e13b3683529287b994ec100ce2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fb69a2208fd425e13b3683529287b994ec100ce2"}]},"branch":"refs/heads/master"},"16344c8345f1c6a39299f78c7b3d8d562c05de78":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-16 06:37:31.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/18/720318/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/18/720318/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/720318/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/720318/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/720318/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/18/720318/3"}}},"commit":{"parents":[{"commit":"782815a9f5c00373923d6f789884a553f5c6e79b","subject":"Add ubuntu-bionic-plain to all regions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/782815a9f5c00373923d6f789884a553f5c6e79b"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-16 01:05:03.000000000","tz":600},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-04-16 06:37:31.000000000","tz":0},"subject":"nodepool: Add more plain images","message":"nodepool: Add more plain images\n\nAdd more images without pip-and-virtualenv we can test on.  Build all\n-plain images on nb04 (the containerised builder) as an additional\ntest of that environment (the existing bionic-plain build is moved\nhere too).\n\nChange-Id: Ic29a25527e259a61ca025325b74acbe64a71774c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/16344c8345f1c6a39299f78c7b3d8d562c05de78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/16344c8345f1c6a39299f78c7b3d8d562c05de78"}]},"branch":"refs/heads/master"},"a3f20a13471f1182ee98647e0e2fdd6a6dba5759":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-04-16 22:42:01.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/18/720318/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/18/720318/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/720318/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/720318/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/720318/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/18/720318/4"}}},"commit":{"parents":[{"commit":"90d6de37acabd647d24c966620b81466a94b9535","subject":"Add ubuntu-bionic-plain to all regions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/90d6de37acabd647d24c966620b81466a94b9535"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-16 01:05:03.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-16 22:41:31.000000000","tz":600},"subject":"nodepool: Add more plain images","message":"nodepool: Add more plain images\n\nAdd more images without pip-and-virtualenv we can test on.  Build all\n-plain images on nb04 (the containerised builder) as an additional\ntest of that environment (the existing bionic-plain build is moved\nhere too).\n\nChange-Id: Ic29a25527e259a61ca025325b74acbe64a71774c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a3f20a13471f1182ee98647e0e2fdd6a6dba5759"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a3f20a13471f1182ee98647e0e2fdd6a6dba5759"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
