)]}'
{"id":"openstack%2Fproject-config~720619","triplet_id":"openstack%2Fproject-config~master~I98b087194c85b73120bb2c2262d4f72d96353d54","project":"openstack/project-config","branch":"master","topic":"centos-arm64","hashtags":[],"change_id":"I98b087194c85b73120bb2c2262d4f72d96353d54","subject":"Add centos aarch64 to labels, unpause","status":"MERGED","created":"2020-04-17 00:00:17.000000000","updated":"2020-04-17 01:28:50.000000000","submitted":"2020-04-17 01:23:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720619-1587086633018-b16e9fe7","meta_rev_id":"9d4e4c9243f83eb41ebde4ee2aaaa61ac9894b2d","_number":720619,"virtual_id_number":720619,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-17 01:23:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-17 01:12:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-17 01:12:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-17 01:23:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"92e2d1a90e76cceb4a4fec9bb3df788a908f77bd","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-04-17 00:00:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41407142e53c78709ed9b99342d7c3132ff466df","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-04-17 00:25:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f1699a0d99aecb819c52b9f01cd5d806e4cd95a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 01:00:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/31e944021cbb42228dcad303492cac77 : SUCCESS in 8m 46s","accounts_in_message":[],"_revision_number":2},{"id":"ae6aee011f69de3a0d60dbd12c78e85e8d33b9f3","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-04-17 01:12:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\ncommitting as trivial.  i have applied manually and built the image, so we want this in sync","accounts_in_message":[],"_revision_number":2},{"id":"e903ea0008f11f149212d353013c4302739fffcd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 01:13:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"961cf974a8680efc277b4bd521fcff12ad9876e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 01:23:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2a1bb948d3f048b0b7f8601c1d6e9abc : SUCCESS in 8m 34s","accounts_in_message":[],"_revision_number":2},{"id":"e78a54c0a95cf2c53669c497f88a87c9a646437d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 01:23:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"88279a0ec2b742cb9c282646129da0ec06f65710","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 01:28:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/5935e10dfd934b618ba94424d36a5ca0 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"768c87448fdfff83845168dd5201d6a5bf3e1d74","revisions":{"db1c8eb0c6359fccb01e463ed0aab0f12bf92274":{"kind":"REWORK","_number":1,"created":"2020-04-17 00:00:17.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/19/720619/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/720619/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/720619/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/720619/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/720619/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/720619/1"}}},"commit":{"parents":[{"commit":"a2bcee7db7114b297504a3c6c9cf4dcdc4bc89a1","subject":"Merge \"nodepool: use job inheritance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a2bcee7db7114b297504a3c6c9cf4dcdc4bc89a1"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-16 23:59:25.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-16 23:59:25.000000000","tz":600},"subject":"Add centos aarch64 to labels","message":"Add centos aarch64 to labels\n\nMissed in I7956efd02859b252b9e327e121b6100b095067ce\n\nChange-Id: I98b087194c85b73120bb2c2262d4f72d96353d54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/db1c8eb0c6359fccb01e463ed0aab0f12bf92274"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/db1c8eb0c6359fccb01e463ed0aab0f12bf92274"}]},"branch":"refs/heads/master"},"768c87448fdfff83845168dd5201d6a5bf3e1d74":{"kind":"REWORK","_number":2,"created":"2020-04-17 00:25:30.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/19/720619/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/720619/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/720619/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/720619/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/720619/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/720619/2"}}},"commit":{"parents":[{"commit":"a2bcee7db7114b297504a3c6c9cf4dcdc4bc89a1","subject":"Merge \"nodepool: use job inheritance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a2bcee7db7114b297504a3c6c9cf4dcdc4bc89a1"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-16 23:59:25.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-17 00:25:09.000000000","tz":600},"subject":"Add centos aarch64 to labels, unpause","message":"Add centos aarch64 to labels, unpause\n\nMissed in I7956efd02859b252b9e327e121b6100b095067ce\n\nChange-Id: I98b087194c85b73120bb2c2262d4f72d96353d54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/768c87448fdfff83845168dd5201d6a5bf3e1d74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/768c87448fdfff83845168dd5201d6a5bf3e1d74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
