)]}'
{"id":"openstack%2Fproject-config~720901","triplet_id":"openstack%2Fproject-config~master~I3e3ab78f87cd5d728237eb8bc6dfc56b2275e868","project":"openstack/project-config","branch":"master","topic":"retire-opendev-puppet","hashtags":[],"change_id":"I3e3ab78f87cd5d728237eb8bc6dfc56b2275e868","subject":"Finish retiring unused puppet modules","status":"MERGED","created":"2020-04-18 15:31:20.000000000","updated":"2020-06-05 19:56:42.000000000","submitted":"2020-06-05 19:50:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720901-1591386636216-0e537f8a","meta_rev_id":"bcf4a89dccf95130592b3d3aa4081062ed24e918","_number":720901,"virtual_id_number":720901,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-05 19:50:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2020-06-05 19:09:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-05 15:59:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2020-06-05 19:10:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":1,"date":"2020-06-05 19:41:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-05 15:59:23.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-06-05 19:10:07.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2020-06-05 19:41:30.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-06-05 19:50:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"04de1389bcea5e1add0fb223963683dfd234f3b1","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-04-18 15:31:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9912c346d265a43f319a2c9973ead41c66d2670c","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-04-18 15:55:48.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"8a45e3fba9e639cd0d52028b72fbc3294c149e08","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-18 16:06:45.000000000","message":"Patch Set 2: Code-Review-1\n\nMissing chagne for zuul.d/projects.yaml","accounts_in_message":[],"_revision_number":2},{"id":"07acb8086f451d4415a4cb12743ce97685cb69a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 16:31:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3d2f1e275da24538a809e5931f7ada36 : SUCCESS in 6m 46s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b926a81cc81f4f1b987345e873cb7ef7 : SUCCESS in 4m 12s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/cfff893fd7d24899bcdc7c7c3f9555e7 : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":2},{"id":"e5025270d9f605b120dce926e80910d8f8473a4a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-21 17:21:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"35664a7fd39980eb545ef4613d0f02e4b765478c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-21 17:30:22.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"85d483c7d6bde7cbd5b874a7a10ad98c6e1cfc01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 17:31:04.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"99400ca9bf02e0b032f81f63150a6c43273ecfbb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-05 14:48:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"92f6d011552bf90cc4bf96788f8c5d6e679c425d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-05 14:49:33.000000000","message":"Patch Set 5: Workflow-1\n\nwaiting for retirement of all repos","accounts_in_message":[],"_revision_number":5},{"id":"5abd6beb8022f7858c730c5d499de1bb7ed60ae2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 15:01:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/42d890c3db164adfb4edc4607a82f539 : SUCCESS in 7m 14s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7dd91f561b4d4eefa5496cec43f8fa47 : SUCCESS in 6m 08s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/de8422fd49d84e0584eb69339488a984 : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":5},{"id":"b6b8239958ae8dbc9c513b2b1fbf1d1257f87130","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-05 15:25:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"97428e681f80c964e70c8fd281b265d7e284bea6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-05 15:26:19.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"787d303a287d290f050429cd6d47ab5cd63a5ca3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 15:30:29.000000000","message":"Patch Set 7: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"2f102d4e4a53ed5bd505054f441e2842c08462ad","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-05 15:45:12.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9e80f9197dbeb43eee435d554ef4a614672a26a1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-05 15:45:57.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"3ca1cdbdacd939a5ad9f6896e7747ff0f1cea99f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-06-05 15:59:23.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"b1fb6e836592f428ca7628495de2a40e76001921","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 16:05:32.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8c1a1dcb98fc42739a7513bbab201b26 : SUCCESS in 8m 01s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/41e06f67be4d4e15b49919b27d910fe7 : SUCCESS in 5m 13s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/7edc141f872a487ca96684283c00929a : SUCCESS in 11m 22s","accounts_in_message":[],"_revision_number":9},{"id":"d15d40db774cae5c9a219569b7e4d67dbad614c3","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-06-05 19:09:04.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"45008436d3694d83fd7b79b5ce4518ac99364730","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-06-05 19:10:01.000000000","message":"Patch Set 9: -Workflow","accounts_in_message":[],"_revision_number":9},{"id":"2a9df461552e411e109774b9f3fa055536b2f3fa","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-06-05 19:10:07.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"ef279c0a46d2ccea0831a42740e6feafb8f70859","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 19:10:23.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"6694bebae8ec699d55b4c6a904186c07ce54f5b8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-05 19:13:45.000000000","message":"Patch Set 9: Workflow-1\n\nstill not everything abandoned","accounts_in_message":[],"_revision_number":9},{"id":"2b5a996d8a641a8c5f4882f453e3f8a1128932ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 19:22:56.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e6b575fd11a642038ffb5f23a6a46a54 : SUCCESS in 7m 15s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/37498ecdda034d80afafb819be5894ba : SUCCESS in 4m 33s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/cf9979e4eee84d24a0657c955f78cf68 : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":9},{"id":"b5481c93891a44030a5a8cead19c2758edf18816","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-05 19:41:30.000000000","message":"Patch Set 9: Workflow+1\n\nall good now","accounts_in_message":[],"_revision_number":9},{"id":"8e61f130dd73868639ac80d7737ee0899ea243d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 19:41:49.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"78fa6a2e461f33a491824565b4a975392bb0d630","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 19:50:35.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b9f7efb7b57c415d8cfb359f7cc1ac9d : SUCCESS in 6m 14s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7eb771da9b9d416f984ee7e2744c6137 : SUCCESS in 4m 59s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/86ffb132afcd48f0a782ba8ce49771ed : SUCCESS in 7m 51s","accounts_in_message":[],"_revision_number":9},{"id":"19af0e87589fc54daef02b5f8fda25092c1f84e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 19:50:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"05e7b45e62a9e1c6fd4d50ac355dda4f49218899","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 19:56:42.000000000","message":"Patch Set 9:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/a68ac87a412449aabecdfe4ffd959020 : SUCCESS in 1m 34s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/fce8f922267d450a9e4fd9154655519b : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"100770eb345f7be10593ea5ed05196a5246df5d3","revisions":{"2daf20a18f2c36d13e967e56f03e788e2ee9c866":{"kind":"REWORK","_number":1,"created":"2020-04-18 15:31:20.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/01/720901/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/720901/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/720901/1"}}},"commit":{"parents":[{"commit":"293293ad0c8c010edc89d225bbb6a7116b0d82e2","subject":"Stop running jobs on unused puppet repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/293293ad0c8c010edc89d225bbb6a7116b0d82e2"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-04-18 15:27:40.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-04-18 15:31:16.000000000","tz":-300},"subject":"Retire unused puppet modules","message":"Retire unused puppet modules\n\nWe don\u0027t use them. Let\u0027s stop caring about them more officially.\n\nChange-Id: I3e3ab78f87cd5d728237eb8bc6dfc56b2275e868\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2daf20a18f2c36d13e967e56f03e788e2ee9c866"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2daf20a18f2c36d13e967e56f03e788e2ee9c866"}]},"branch":"refs/heads/master"},"b43e10a6af2db1319708f2371addc1f50609f98f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-18 15:55:48.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/01/720901/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/720901/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/720901/2"}}},"commit":{"parents":[{"commit":"3c1b53da256a3195d8fbc5f554ec364ad2e8d59c","subject":"Stop running jobs on unused puppet repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3c1b53da256a3195d8fbc5f554ec364ad2e8d59c"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-04-18 15:27:40.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-04-18 15:55:44.000000000","tz":-300},"subject":"Retire unused puppet modules","message":"Retire unused puppet modules\n\nWe don\u0027t use them. Let\u0027s stop caring about them more officially.\n\nChange-Id: I3e3ab78f87cd5d728237eb8bc6dfc56b2275e868\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b43e10a6af2db1319708f2371addc1f50609f98f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b43e10a6af2db1319708f2371addc1f50609f98f"}]},"branch":"refs/heads/master"},"ccfcb96e0e4c5e1d1eee38d24bf00f84de46d952":{"kind":"REWORK","_number":3,"created":"2020-04-21 17:21:55.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/01/720901/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/720901/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/720901/3"}}},"commit":{"parents":[{"commit":"3c1b53da256a3195d8fbc5f554ec364ad2e8d59c","subject":"Stop running jobs on unused puppet repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3c1b53da256a3195d8fbc5f554ec364ad2e8d59c"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-04-18 15:27:40.000000000","tz":-300},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-21 17:21:43.000000000","tz":120},"subject":"Retire unused puppet modules","message":"Retire unused puppet modules\n\nWe don\u0027t use them. Let\u0027s stop caring about them more officially.\n\nChange-Id: I3e3ab78f87cd5d728237eb8bc6dfc56b2275e868\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ccfcb96e0e4c5e1d1eee38d24bf00f84de46d952"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ccfcb96e0e4c5e1d1eee38d24bf00f84de46d952"}]},"branch":"refs/heads/master"},"2fe1833accddb3820b0fd99812eedaf4d644e71b":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-04-21 17:30:22.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/01/720901/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/720901/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/720901/4"}}},"commit":{"parents":[{"commit":"3c1b53da256a3195d8fbc5f554ec364ad2e8d59c","subject":"Stop running jobs on unused puppet repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3c1b53da256a3195d8fbc5f554ec364ad2e8d59c"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-04-18 15:27:40.000000000","tz":-300},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-21 17:29:59.000000000","tz":120},"subject":"Retire unused puppet modules","message":"Retire unused puppet modules\n\nWe don\u0027t use them. Let\u0027s stop caring about them more officially.\n\nDepends-On: https://review.opendev.org/720892\nChange-Id: I3e3ab78f87cd5d728237eb8bc6dfc56b2275e868\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2fe1833accddb3820b0fd99812eedaf4d644e71b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2fe1833accddb3820b0fd99812eedaf4d644e71b"}]},"branch":"refs/heads/master"},"918c603b9f2566bca9491f9a720b35e5768a5bb6":{"kind":"REWORK","_number":5,"created":"2020-06-05 14:48:29.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/01/720901/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/720901/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/720901/5"}}},"commit":{"parents":[{"commit":"6f33aa5cc26140e8028bd616352ce95695e3e124","subject":"Retire more opendev puppet modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6f33aa5cc26140e8028bd616352ce95695e3e124"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-04-18 15:27:40.000000000","tz":-300},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-05 14:48:02.000000000","tz":120},"subject":"Retire unused puppet modules","message":"Retire unused puppet modules\n\nWe don\u0027t use them. Let\u0027s stop caring about them more officially.\n\nNote that some of the initially proposed repos have *not* been retired,\nso their jobs are added back in here. The repos are\nopendev/puppet-{jenkins,nodepool,openstackci,zuul}.\n\nDepends-On: https://review.opendev.org/720892\nDepends-On: https://review.opendev.org/729321\nDepends-On: https://review.opendev.org/731947\nChange-Id: I3e3ab78f87cd5d728237eb8bc6dfc56b2275e868\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/918c603b9f2566bca9491f9a720b35e5768a5bb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/918c603b9f2566bca9491f9a720b35e5768a5bb6"}]},"branch":"refs/heads/master"},"cae296bbcd21de5c6ccc45250c69ced8091bb6a4":{"kind":"REWORK","_number":6,"created":"2020-06-05 15:25:35.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/01/720901/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/720901/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/720901/6"}}},"commit":{"parents":[{"commit":"db7ba972fd24cee9bf50afeccb9e24da757a42bd","subject":"Retire more opendev puppet modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/db7ba972fd24cee9bf50afeccb9e24da757a42bd"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-04-18 15:27:40.000000000","tz":-300},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-05 15:25:11.000000000","tz":120},"subject":"Retire unused puppet modules","message":"Retire unused puppet modules\n\nWe don\u0027t use them. Let\u0027s stop caring about them more officially.\n\nNote that some of the initially proposed repos have *not* been retired,\nso their jobs are added back in here. The repos are\nopendev/puppet-{jenkins,nodepool,openstackci,zuul}.\n\nDepends-On: https://review.opendev.org/720892\nDepends-On: https://review.opendev.org/729321\nDepends-On: https://review.opendev.org/731947\nDepends-On: https://review.opendev.org/733851\nDepends-On: https://review.opendev.org/722018\nDepends-On: https://review.opendev.org/722001\nDepends-On: https://review.opendev.org/733855\nDepends-On: https://review.opendev.org/722002\nDepends-On: https://review.opendev.org/722003\nDepends-On: https://review.opendev.org/722004\nDepends-On: https://review.opendev.org/722005\nDepends-On: https://review.opendev.org/722006\nDepends-On: https://review.opendev.org/722007\nDepends-On: https://review.opendev.org/722008\nDepends-On: https://review.opendev.org/722009\nDepends-On: https://review.opendev.org/733852\nDepends-On: https://review.opendev.org/722013\nDepends-On: https://review.opendev.org/722014\nDepends-On: https://review.opendev.org/722015\nDepends-On: https://review.opendev.org/733853\nChange-Id: I3e3ab78f87cd5d728237eb8bc6dfc56b2275e868\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cae296bbcd21de5c6ccc45250c69ced8091bb6a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cae296bbcd21de5c6ccc45250c69ced8091bb6a4"}]},"branch":"refs/heads/master"},"365549134927b1d832ec87a1f7da6bd91727a710":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2020-06-05 15:26:19.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/01/720901/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/720901/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/720901/7"}}},"commit":{"parents":[{"commit":"db7ba972fd24cee9bf50afeccb9e24da757a42bd","subject":"Retire more opendev puppet modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/db7ba972fd24cee9bf50afeccb9e24da757a42bd"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-04-18 15:27:40.000000000","tz":-300},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-05 15:26:03.000000000","tz":120},"subject":"Finish retiring unused puppet modules","message":"Finish retiring unused puppet modules\n\nWe don\u0027t use them. Let\u0027s stop caring about them more officially.\n\nNote that some of the initially proposed repos have *not* been retired,\nso their jobs are added back in here. The repos are\nopendev/puppet-{jenkins,nodepool,openstackci,zuul}.\n\nDepends-On: https://review.opendev.org/720892\nDepends-On: https://review.opendev.org/729321\nDepends-On: https://review.opendev.org/731947\nDepends-On: https://review.opendev.org/733851\nDepends-On: https://review.opendev.org/722018\nDepends-On: https://review.opendev.org/722001\nDepends-On: https://review.opendev.org/733855\nDepends-On: https://review.opendev.org/722002\nDepends-On: https://review.opendev.org/722003\nDepends-On: https://review.opendev.org/722004\nDepends-On: https://review.opendev.org/722005\nDepends-On: https://review.opendev.org/722006\nDepends-On: https://review.opendev.org/722007\nDepends-On: https://review.opendev.org/722008\nDepends-On: https://review.opendev.org/722009\nDepends-On: https://review.opendev.org/733852\nDepends-On: https://review.opendev.org/722013\nDepends-On: https://review.opendev.org/722014\nDepends-On: https://review.opendev.org/722015\nDepends-On: https://review.opendev.org/733853\nChange-Id: I3e3ab78f87cd5d728237eb8bc6dfc56b2275e868\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/365549134927b1d832ec87a1f7da6bd91727a710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/365549134927b1d832ec87a1f7da6bd91727a710"}]},"branch":"refs/heads/master"},"f0f6e1205f72c70fe0935c413f535b4c086e8e2a":{"kind":"REWORK","_number":8,"created":"2020-06-05 15:45:12.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/01/720901/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/720901/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/720901/8"}}},"commit":{"parents":[{"commit":"1baebb41a3ea65e5a4c31bce2b2601e2ec30cdb8","subject":"Retire more opendev puppet modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1baebb41a3ea65e5a4c31bce2b2601e2ec30cdb8"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-04-18 15:27:40.000000000","tz":-300},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-05 15:44:33.000000000","tz":120},"subject":"Finish retiring unused puppet modules","message":"Finish retiring unused puppet modules\n\nWe don\u0027t use them. Let\u0027s stop caring about them more officially.\n\nThis retires the following repositories:\n    opendev/puppet-accessbot\n    opendev/puppet-ansible\n    opendev/puppet-cgit\n    opendev/puppet-drupal\n    opendev/puppet-etherpad_lite\n    opendev/puppet-exim\n    opendev/puppet-gerrit\n    opendev/puppet-gerritbot\n    opendev/puppet-github\n    opendev/puppet-infracloud\n    opendev/puppet-ipsilon\n    opendev/puppet-iptables\n    opendev/puppet-os_client_config\n    opendev/puppet-packagekit\n    opendev/puppet-phabricator\n    opendev/puppet-snmpd\n\nNote that some of the initially proposed repos have *not* been retired,\nso their jobs are added back in here. The repos are\nopendev/puppet-{jenkins,nodepool,openstackci,zuul}.\n\nDepends-On: https://review.opendev.org/720892\nDepends-On: https://review.opendev.org/729321\nDepends-On: https://review.opendev.org/731947\nDepends-On: https://review.opendev.org/733851\nDepends-On: https://review.opendev.org/722018\nDepends-On: https://review.opendev.org/722001\nDepends-On: https://review.opendev.org/733855\nDepends-On: https://review.opendev.org/722002\nDepends-On: https://review.opendev.org/722003\nDepends-On: https://review.opendev.org/722004\nDepends-On: https://review.opendev.org/722005\nDepends-On: https://review.opendev.org/722006\nDepends-On: https://review.opendev.org/722007\nDepends-On: https://review.opendev.org/722008\nDepends-On: https://review.opendev.org/722009\nDepends-On: https://review.opendev.org/722013\nDepends-On: https://review.opendev.org/722014\nDepends-On: https://review.opendev.org/722015\nDepends-On: https://review.opendev.org/733853\nChange-Id: I3e3ab78f87cd5d728237eb8bc6dfc56b2275e868\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f0f6e1205f72c70fe0935c413f535b4c086e8e2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f0f6e1205f72c70fe0935c413f535b4c086e8e2a"}]},"branch":"refs/heads/master"},"100770eb345f7be10593ea5ed05196a5246df5d3":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2020-06-05 15:45:57.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/01/720901/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/720901/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/720901/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/720901/9"}}},"commit":{"parents":[{"commit":"1baebb41a3ea65e5a4c31bce2b2601e2ec30cdb8","subject":"Retire more opendev puppet modules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1baebb41a3ea65e5a4c31bce2b2601e2ec30cdb8"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-04-18 15:27:40.000000000","tz":-300},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-05 15:45:30.000000000","tz":120},"subject":"Finish retiring unused puppet modules","message":"Finish retiring unused puppet modules\n\nWe don\u0027t use them. Let\u0027s stop caring about them more officially.\n\nThis retires the following repositories:\n    opendev/puppet-accessbot\n    opendev/puppet-ansible\n    opendev/puppet-cgit\n    opendev/puppet-drupal\n    opendev/puppet-etherpad_lite\n    opendev/puppet-exim\n    opendev/puppet-gerrit\n    opendev/puppet-gerritbot\n    opendev/puppet-github\n    opendev/puppet-infracloud\n    opendev/puppet-ipsilon\n    opendev/puppet-iptables\n    opendev/puppet-os_client_config\n    opendev/puppet-packagekit\n    opendev/puppet-phabricator\n    opendev/puppet-snmpd\n\nNote that some of the initially proposed repos have *not* been retired,\nso their jobs are added back in here. The repos are\nopendev/puppet-{jenkins,nodepool,openstackci,zuul}.\n\nDepends-On: https://review.opendev.org/720892\nDepends-On: https://review.opendev.org/729321\nDepends-On: https://review.opendev.org/731947\nDepends-On: https://review.opendev.org/733851\nDepends-On: https://review.opendev.org/722018\nDepends-On: https://review.opendev.org/722001\nDepends-On: https://review.opendev.org/733855\nDepends-On: https://review.opendev.org/722002\nDepends-On: https://review.opendev.org/722003\nDepends-On: https://review.opendev.org/722004\nDepends-On: https://review.opendev.org/722005\nDepends-On: https://review.opendev.org/722006\nDepends-On: https://review.opendev.org/722007\nDepends-On: https://review.opendev.org/722008\nDepends-On: https://review.opendev.org/722009\nDepends-On: https://review.opendev.org/722013\nDepends-On: https://review.opendev.org/722014\nDepends-On: https://review.opendev.org/722015\nDepends-On: https://review.opendev.org/733853\nCo-Authored-By: Andreas Jaeger \u003caj@suse.com\u003e\nChange-Id: I3e3ab78f87cd5d728237eb8bc6dfc56b2275e868\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/100770eb345f7be10593ea5ed05196a5246df5d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/100770eb345f7be10593ea5ed05196a5246df5d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
