)]}'
{"id":"openstack%2Fproject-config~721159","triplet_id":"openstack%2Fproject-config~master~I243eccbc6ee412cf1ff29979dd0fdcc4d577d3c8","project":"openstack/project-config","branch":"master","topic":"f31-fix","hashtags":[],"change_id":"I243eccbc6ee412cf1ff29979dd0fdcc4d577d3c8","subject":"Fix release for Fedora 31","status":"MERGED","created":"2020-04-20 03:25:55.000000000","updated":"2020-04-20 03:52:35.000000000","submitted":"2020-04-20 03:47:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721159-1587354458563-b8f12ef2","meta_rev_id":"10ea49b5a2f8bff3aa538b5ff324956936a04f96","_number":721159,"virtual_id_number":721159,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-20 03:47:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-20 03:35:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-20 03:35:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-20 03:47:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dddde8a77a775a13b02ccb2cdfcd6a001a7a1a1e","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-04-20 03:25:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"577b9276c287c95729bfea92be8c95ff752b1fb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 03:32:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c691a8b814e9412e872bc1223f55570c : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":1},{"id":"36e5692cc71bbecce6a79ae569a9095ecae90a13","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-04-20 03:35:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\ncommitting as trivial","accounts_in_message":[],"_revision_number":1},{"id":"263134117eb89cdd311776b08ab0beb22de0b645","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 03:36:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1194cc038f9f70caec2c43a91e8fc0c9c9b26769","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 03:47:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f80b635a679642019f15a27f366c9fcf : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":1},{"id":"1b99bed8503d575cb63844cb0e918e749fb9b986","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 03:47:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f97c74aae88c545226cf90388ed9da9a48e75630","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 03:52:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/b5cef942b0e646e58e6141535a1f598c : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"054a73ea3e457de67181ce3cef1dd89ac8456668","revisions":{"054a73ea3e457de67181ce3cef1dd89ac8456668":{"kind":"REWORK","_number":1,"created":"2020-04-20 03:25:55.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/59/721159/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/721159/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/721159/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/721159/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/721159/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/721159/1"}}},"commit":{"parents":[{"commit":"3d12df3a9cfbd5c20c7c12f2f538a2aa41fd3e2e","subject":"Merge \"Change gerrit ACLs for cinder-tempest-plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3d12df3a9cfbd5c20c7c12f2f538a2aa41fd3e2e"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-20 03:25:19.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-04-20 03:25:19.000000000","tz":600},"subject":"Fix release for Fedora 31","message":"Fix release for Fedora 31\n\nWe accidentally set this back to 30 in the reorg\n\nChange-Id: I243eccbc6ee412cf1ff29979dd0fdcc4d577d3c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/054a73ea3e457de67181ce3cef1dd89ac8456668"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/054a73ea3e457de67181ce3cef1dd89ac8456668"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
