)]}'
{"id":"openstack%2Fproject-config~721328","triplet_id":"openstack%2Fproject-config~master~I703ac705b549f4f4a7bc24d37a63b88035b5e892","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I703ac705b549f4f4a7bc24d37a63b88035b5e892","subject":"Add Airship subproject documentation job","status":"MERGED","created":"2020-04-20 18:55:26.000000000","updated":"2020-04-27 06:04:25.000000000","submitted":"2020-04-27 06:04:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721328-1587967465541-a0c18567","meta_rev_id":"3fc63a95d069062896b06d8d3012d1ed450697e9","_number":721328,"virtual_id_number":721328,"owner":{"_account_id":28618,"name":"Drew Walters","email":"drewwalters@microsoft.com","username":"drewwalters96"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28618,"name":"Drew Walters","email":"drewwalters@microsoft.com","username":"drewwalters96"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 06:04:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-04-27 05:51:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28618,"name":"Drew Walters","email":"drewwalters@microsoft.com","username":"drewwalters96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-27 04:50:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2020-04-27 05:51:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28618,"name":"Drew Walters","email":"drewwalters@microsoft.com","username":"drewwalters96"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28618,"name":"Drew Walters","email":"drewwalters@microsoft.com","username":"drewwalters96"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-27 04:50:46.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-04-27 05:51:25.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-27 06:04:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c11c07af6a05171c68a7b134baaa8d9ec847f51d","author":{"_account_id":28618,"name":"Drew Walters","email":"drewwalters@microsoft.com","username":"drewwalters96"},"date":"2020-04-20 18:55:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01cc83b07aea8eb3d8c0d8ae6c6ed8ee7549c465","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-20 19:00:58.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"fcc8ff1c28acfd737b9259e873d2c209a37d99dc","author":{"_account_id":28618,"name":"Drew Walters","email":"drewwalters@microsoft.com","username":"drewwalters96"},"date":"2020-04-20 19:08:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e63d0420002c7a61c54ca42105a1cb019c258853","author":{"_account_id":28618,"name":"Drew Walters","email":"drewwalters@microsoft.com","username":"drewwalters96"},"date":"2020-04-20 19:08:47.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4a223b8b4a6ee63d156a1ecded0d22ae42086ffb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-20 19:16:11.000000000","message":"Patch Set 2:\n\nLGTM - I like Ian to review, I pinged him on IRC.\n\nWill wait with voting a bit.","accounts_in_message":[],"_revision_number":2},{"id":"bd0c9a10c157451514db4a9c77b64da04fc48634","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 19:25:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4880fc68f4ea47e5aaae54bf808da00c : SUCCESS in 9m 36s","accounts_in_message":[],"_revision_number":2},{"id":"cd721673ab28cad6e3bb66595bdc593c186afb8a","author":{"_account_id":28618,"name":"Drew Walters","email":"drewwalters@microsoft.com","username":"drewwalters96"},"date":"2020-04-24 14:30:48.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c6ac6c0edbd85473130f218ad8a04b45cae29edc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 14:41:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/593e939bf6004a1eb2f2fc011fea61c3 : SUCCESS in 8m 50s","accounts_in_message":[],"_revision_number":3},{"id":"d7b030bc193ae4cfc05a2acd35fd2b62d7fd8f26","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-24 14:50:00.000000000","message":"Patch Set 3:\n\nDrew, why was this rebased?","accounts_in_message":[],"_revision_number":3},{"id":"e2ab8fc23e527453882018850b69798c11e7bc7e","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-04-27 04:50:46.000000000","message":"Patch Set 3: Code-Review+2\n\nthis should publish ok I think; i don\u0027t think the sites will overwrite each other","accounts_in_message":[],"_revision_number":3},{"id":"287734e5fd8c681d78875bedd34728049922a804","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-27 05:51:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d59e764baa7a9c86a0f42eae8a505bf07f1f3910","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 05:51:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5e2e264abb216e23cedcf162c7528f9daffc269a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 06:04:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/61ed6675d63a4e1e9d303239328a1415 : SUCCESS in 7m 30s","accounts_in_message":[],"_revision_number":3},{"id":"3fc63a95d069062896b06d8d3012d1ed450697e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 06:04:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cc8c5d7083ea9db9b618044564d8302757cb6a16","revisions":{"5c277410809fe388f4e4c4535fb1c1ff9a6ae87d":{"kind":"REWORK","_number":1,"created":"2020-04-20 18:55:26.000000000","uploader":{"_account_id":28618,"name":"Drew Walters","email":"drewwalters@microsoft.com","username":"drewwalters96"},"ref":"refs/changes/28/721328/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/28/721328/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/28/721328/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/28/721328/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/28/721328/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/28/721328/1"}}},"commit":{"parents":[{"commit":"58ea0b53c245a2d6a813631c813a4b4f167a046a","subject":"Merge \"Remove pypy job from x/surveil\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/58ea0b53c245a2d6a813631c813a4b4f167a046a"}]}],"author":{"name":"Drew Walters","email":"andrew.walters@att.com","date":"2020-04-20 16:57:10.000000000","tz":0},"committer":{"name":"Drew Walters","email":"andrew.walters@att.com","date":"2020-04-20 18:54:55.000000000","tz":0},"subject":"Add Airship subproject documentation job","message":"Add Airship subproject documentation job\n\nThe airship/treasuremap project publishes to docs.airshipit.org, but\nother airship repositories cannot publish to the same site. This change\nadds a separate job that enables other Airship repositories to promote\ntheir documentation to a subdirectory of the main documentation site.\nThis change also replaces airship/treasuremap with airship/docs as the\nmain documentation repository.\n\nChange-Id: I703ac705b549f4f4a7bc24d37a63b88035b5e892\nSigned-off-by: Drew Walters \u003candrew.walters@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5c277410809fe388f4e4c4535fb1c1ff9a6ae87d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5c277410809fe388f4e4c4535fb1c1ff9a6ae87d"}]},"branch":"refs/heads/master"},"dbc063b98dcbd0e0fd97a4f92da6537aa452b2ff":{"kind":"REWORK","_number":2,"created":"2020-04-20 19:08:39.000000000","uploader":{"_account_id":28618,"name":"Drew Walters","email":"drewwalters@microsoft.com","username":"drewwalters96"},"ref":"refs/changes/28/721328/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/28/721328/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/28/721328/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/28/721328/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/28/721328/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/28/721328/2"}}},"commit":{"parents":[{"commit":"58ea0b53c245a2d6a813631c813a4b4f167a046a","subject":"Merge \"Remove pypy job from x/surveil\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/58ea0b53c245a2d6a813631c813a4b4f167a046a"}]}],"author":{"name":"Drew Walters","email":"andrew.walters@att.com","date":"2020-04-20 16:57:10.000000000","tz":0},"committer":{"name":"Drew Walters","email":"andrew.walters@att.com","date":"2020-04-20 19:08:06.000000000","tz":0},"subject":"Add Airship subproject documentation job","message":"Add Airship subproject documentation job\n\nThe airship/treasuremap project publishes to docs.airshipit.org, but\nother airship repositories cannot publish to the same site. This change\nadds a separate job that enables other Airship repositories to promote\ntheir documentation to a subdirectory of the main documentation site.\nThis change also replaces airship/treasuremap with airship/docs as the\nmain documentation repository.\n\nChange-Id: I703ac705b549f4f4a7bc24d37a63b88035b5e892\nSigned-off-by: Drew Walters \u003candrew.walters@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dbc063b98dcbd0e0fd97a4f92da6537aa452b2ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dbc063b98dcbd0e0fd97a4f92da6537aa452b2ff"}]},"branch":"refs/heads/master"},"cc8c5d7083ea9db9b618044564d8302757cb6a16":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-24 14:30:48.000000000","uploader":{"_account_id":28618,"name":"Drew Walters","email":"drewwalters@microsoft.com","username":"drewwalters96"},"ref":"refs/changes/28/721328/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/28/721328/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/28/721328/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/28/721328/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/28/721328/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/28/721328/3"}}},"commit":{"parents":[{"commit":"f1899bd92d6cf1d40918cbd73a450146acb295b6","subject":"Merge \"Finish retiring i18n-specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f1899bd92d6cf1d40918cbd73a450146acb295b6"}]}],"author":{"name":"Drew Walters","email":"andrew.walters@att.com","date":"2020-04-20 16:57:10.000000000","tz":0},"committer":{"name":"Drew Walters","email":"andrew.walters@att.com","date":"2020-04-24 14:30:48.000000000","tz":0},"subject":"Add Airship subproject documentation job","message":"Add Airship subproject documentation job\n\nThe airship/treasuremap project publishes to docs.airshipit.org, but\nother airship repositories cannot publish to the same site. This change\nadds a separate job that enables other Airship repositories to promote\ntheir documentation to a subdirectory of the main documentation site.\nThis change also replaces airship/treasuremap with airship/docs as the\nmain documentation repository.\n\nChange-Id: I703ac705b549f4f4a7bc24d37a63b88035b5e892\nSigned-off-by: Drew Walters \u003candrew.walters@att.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cc8c5d7083ea9db9b618044564d8302757cb6a16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cc8c5d7083ea9db9b618044564d8302757cb6a16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
