)]}'
{"id":"openstack%2Fproject-config~721565","triplet_id":"openstack%2Fproject-config~master~Idaa872b6cb226d3b28b8f506544ffcc044528ddd","project":"openstack/project-config","branch":"master","topic":"retire-networking-ovn","hashtags":[],"change_id":"Idaa872b6cb226d3b28b8f506544ffcc044528ddd","subject":"Retire openstack/networking-ovn repository","status":"MERGED","created":"2020-04-21 10:18:51.000000000","updated":"2020-04-25 01:06:21.000000000","submitted":"2020-04-24 19:04:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721565-1587755081474-602cea05","meta_rev_id":"3a717836f5e5aa11d1f0427ba204cdeb10e6787c","_number":721565,"virtual_id_number":721565,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"date":"2020-04-24 16:11:57.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-24 19:04:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-04-24 18:52:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-04-24 18:37:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2020-04-24 18:52:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-21 13:01:55.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-04-24 07:58:49.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-04-24 16:11:57.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-04-24 18:37:59.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-24 18:52:28.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-04-24 19:04:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8ef15c3ff56bcf17e58111aa9d0b3f385dbfc6eb","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-21 10:18:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"033e3a9487526994c9daabb5d40274eec772c472","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-21 10:19:04.000000000","message":"Topic set to retire-networking-ovn","accounts_in_message":[],"_revision_number":1},{"id":"fe79bc146fcfea4f2cfd78594a8417b2a1a2d50c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 10:26:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/80472e0ca3a047afa0fb2d7c20747a18 : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":1},{"id":"10d8b239c94dd3eea983fb55c2e0061915c0e217","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-21 10:28:29.000000000","message":"Patch Set 1: Code-Review+1\n\nNeutron BP is now completed and all code is moved to neutron repo already.","accounts_in_message":[],"_revision_number":1},{"id":"4d77621659974530384e49f263317e9e9466d9e5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-21 10:32:31.000000000","message":"Patch Set 1:\n\nDo you want to stop work on master and keeping stable branches around?\n\nWe understand with Retirement a full retirement with completely making the repo read-only, see https://docs.openstack.org/infra/manual/drivers.html#retiring-a-project\n\nSo, either follow the steps above - or make clear what you want in the commit message","accounts_in_message":[],"_revision_number":1},{"id":"87ccba09e01091d9e6a9a255b07c30afb4ab1f47","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-21 10:55:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5f142f5fa5e0564313235ff761c4534280691793","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-21 10:56:58.000000000","message":"Patch Set 1:\n\n\u003e Do you want to stop work on master and keeping stable branches\n \u003e around?\n \u003e \n \u003e We understand with Retirement a full retirement with completely\n \u003e making the repo read-only, see https://docs.openstack.org/infra/manual/drivers.html#retiring-a-project\n \u003e \n \u003e So, either follow the steps above - or make clear what you want in\n \u003e the commit message\n\nWe will keep stable branches and accept bug fixes. I modified the project.yaml to correspond that.","accounts_in_message":[],"_revision_number":1},{"id":"d729a7d5342176352d6a40fe7e97eaa397a23421","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 11:03:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f171a8bcaadf4c7f955bed1fb7989689 : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":2},{"id":"b0cf5ed83aff5646a855899a1c8971b37eaddf9f","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-04-21 13:01:55.000000000","message":"Patch Set 2: Code-Review+1\n\nWill you be removing all the content from master except for README and .gitreview as specified in [1]?  If so, I would say maybe wait until after the Ussuri release and remove all content from master first thing in Victoria.\n\n[1] https://docs.opendev.org/opendev/infra-manual/latest/drivers.html#step-2-remove-project-content","accounts_in_message":[],"_revision_number":2},{"id":"787c3ca4b619171e6875e24cce6c3a066ce481e7","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-21 13:28:19.000000000","message":"Patch Set 2:\n\n\u003e Will you be removing all the content from master except for README\n \u003e and .gitreview as specified in [1]?  If so, I would say maybe wait\n \u003e until after the Ussuri release and remove all content from master\n \u003e first thing in Victoria.\n \u003e \n \u003e [1] https://docs.opendev.org/opendev/infra-manual/latest/drivers.html#step-2-remove-project-content\n\nHey!\n\nYes. The goal is to clean master branch and leave proper README note along with .gitreview.\nThe last stable branch in networking-ovn is and will still be \u0027stable/train\u0027. \n\nThe \u0027stable/ussuri\u0027 code will be released under Neutron repository.\n\nI\u0027m ok with waiting for Ussuri release first.","accounts_in_message":[],"_revision_number":2},{"id":"2e992b1e28b46bf9a69c15bbdfc8064def21ddc1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-04-24 07:58:49.000000000","message":"Patch Set 2: Code-Review+1\n\n\u003e \u003e Will you be removing all the content from master except for\n \u003e README\n \u003e \u003e and .gitreview as specified in [1]?  If so, I would say maybe\n \u003e wait\n \u003e \u003e until after the Ussuri release and remove all content from master\n \u003e \u003e first thing in Victoria.\n \u003e \u003e\n \u003e \u003e [1] https://docs.opendev.org/opendev/infra-manual/latest/drivers.html#step-2-remove-project-content\n \u003e \n \u003e Hey!\n \u003e \n \u003e Yes. The goal is to clean master branch and leave proper README\n \u003e note along with .gitreview.\n \u003e The last stable branch in networking-ovn is and will still be\n \u003e \u0027stable/train\u0027.\n \u003e \n \u003e The \u0027stable/ussuri\u0027 code will be released under Neutron repository.\n \u003e \n \u003e I\u0027m ok with waiting for Ussuri release first.\n\nI\u0027m not sure if this wait is really needed. We will not have Ussuri release of networking-ovn anymore and we will not cut stable/ussuri branch for it.","accounts_in_message":[],"_revision_number":2},{"id":"327a321f9baa7d9ced10d7dc6599a89fb1e43c07","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-24 15:15:27.000000000","message":"Patch Set 2: Code-Review-1\n\nRemove option: translate from gerrit/projects.yaml as well.","accounts_in_message":[],"_revision_number":2},{"id":"2c7c4451d878b122c7c9c2d0902de5741f9c7767","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-24 16:10:51.000000000","message":"Patch Set 2:\n\nI\u0027ll address Andreas\u0027 response to keep this moving along.","accounts_in_message":[],"_revision_number":2},{"id":"ff8f3c80a9011fb6861478a031e25cb52408247a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-24 16:10:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6710ffcb4768e37f22d117139d13876839b2ebe1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-24 16:11:57.000000000","message":"Patch Set 3:\n\nI\u0027ll also send a follow-on to remove the grafana dashboard as it has no data any more.","accounts_in_message":[],"_revision_number":3},{"id":"5349fb7f61c3e2017917012d76e2abaf9c91e25b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 16:27:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4dd3da776cf240acae948e18f9df4a2d : SUCCESS in 9m 57s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/928e441b80064b64951345663b6882b6 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":3},{"id":"ebdd0e504b7276f0c3e524f563e8b8fe2058b004","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-24 18:37:59.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b2bcd8a40aca1cb3d69ba7253a611b5c36cf2674","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-04-24 18:52:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8ecd191e9c9aa8eb9727848507887376f0c6cd50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 18:52:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bf9576160325445f3085978896d0a52fa2020eec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 19:04:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e84a5cd34368e59a9a132cf70e8a7f0ac88ed564","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 19:04:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ea2b1c1b8c244a93a64c2d040b898743 : SUCCESS in 8m 55s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c2f8c2af199841bf84eeae0b0bcf6a06 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":3},{"id":"3a717836f5e5aa11d1f0427ba204cdeb10e6787c","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 01:06:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/b74889d8bc75487d8ed419f9a0cf1be0 : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d3181bdfb4a9c13e5ac40397ddffefcfbc9a2480","revisions":{"8544b1c7d3162c997f68edb53471004a6c102ede":{"kind":"REWORK","_number":1,"created":"2020-04-21 10:18:51.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/65/721565/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/65/721565/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/721565/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/721565/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/721565/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/65/721565/1"}}},"commit":{"parents":[{"commit":"67b12b9f0df71f03859a44f978784a51d3aebb73","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/67b12b9f0df71f03859a44f978784a51d3aebb73"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-04-21 10:15:39.000000000","tz":120},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-04-21 10:15:39.000000000","tz":120},"subject":"Retire openstack/networking-ovn repository","message":"Retire openstack/networking-ovn repository\n\nDuring Ussuri cycle code from this repository has been\nmoved to Neutron [1].\n\nStart the retirement of openstack/networking-ovn with adding\nnoop-jobs to the repo.\n\n[1] https://blueprints.launchpad.net/neutron/+spec/neutron-ovn-merge\n\nChange-Id: Idaa872b6cb226d3b28b8f506544ffcc044528ddd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8544b1c7d3162c997f68edb53471004a6c102ede"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8544b1c7d3162c997f68edb53471004a6c102ede"}]},"branch":"refs/heads/master"},"8ecad339bc012d9998ab4fc5f82699d5279028d3":{"kind":"REWORK","_number":2,"created":"2020-04-21 10:55:32.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/65/721565/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/65/721565/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/721565/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/721565/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/721565/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/65/721565/2"}}},"commit":{"parents":[{"commit":"67b12b9f0df71f03859a44f978784a51d3aebb73","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/67b12b9f0df71f03859a44f978784a51d3aebb73"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-04-21 10:15:39.000000000","tz":120},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-04-21 10:56:55.000000000","tz":120},"subject":"Retire openstack/networking-ovn repository","message":"Retire openstack/networking-ovn repository\n\nDuring Ussuri cycle code from this repository has been\nmoved to Neutron [1].\n\nAll changes to master branch will not be accepted, but we will\nkeep stable branches and accept bug fixes.\n\nAnnouncement:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-April/014307.html\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-April/014308.html\n\n[1] https://blueprints.launchpad.net/neutron/+spec/neutron-ovn-merge\n\nChange-Id: Idaa872b6cb226d3b28b8f506544ffcc044528ddd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8ecad339bc012d9998ab4fc5f82699d5279028d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8ecad339bc012d9998ab4fc5f82699d5279028d3"}]},"branch":"refs/heads/master"},"d3181bdfb4a9c13e5ac40397ddffefcfbc9a2480":{"kind":"REWORK","_number":3,"created":"2020-04-24 16:10:58.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/65/721565/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/65/721565/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/721565/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/721565/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/65/721565/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/65/721565/3"}}},"commit":{"parents":[{"commit":"67b12b9f0df71f03859a44f978784a51d3aebb73","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/67b12b9f0df71f03859a44f978784a51d3aebb73"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-04-21 10:15:39.000000000","tz":120},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-24 16:10:22.000000000","tz":-240},"subject":"Retire openstack/networking-ovn repository","message":"Retire openstack/networking-ovn repository\n\nDuring Ussuri cycle code from this repository has been\nmoved to Neutron [1].\n\nAll changes to master branch will not be accepted, but we will\nkeep stable branches and accept bug fixes.\n\nAnnouncement:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-April/014307.html\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-April/014308.html\n\n[1] https://blueprints.launchpad.net/neutron/+spec/neutron-ovn-merge\n\nChange-Id: Idaa872b6cb226d3b28b8f506544ffcc044528ddd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d3181bdfb4a9c13e5ac40397ddffefcfbc9a2480"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d3181bdfb4a9c13e5ac40397ddffefcfbc9a2480"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
