)]}'
{"id":"openstack%2Fproject-config~721722","triplet_id":"openstack%2Fproject-config~master~Iec11018b3a17c7667f91e27856e68e922cae8336","project":"openstack/project-config","branch":"master","topic":"retire-i18n-specs","hashtags":[],"change_id":"Iec11018b3a17c7667f91e27856e68e922cae8336","subject":"Finish retiring i18n-specs","status":"MERGED","created":"2020-04-21 20:10:20.000000000","updated":"2020-04-24 14:08:48.000000000","submitted":"2020-04-24 13:43:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721722-1587735818903-dfe5b5ea","meta_rev_id":"23c826c11e8f2f40141637ae3223acd574b60d38","_number":721722,"virtual_id_number":721722,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-24 13:43:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"recommended":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2020-04-22 15:28:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2020-04-24 13:28:48.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-24 03:06:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2020-04-24 13:28:48.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-22 09:24:13.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2020-04-24 03:06:14.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-04-24 13:28:48.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-04-24 13:43:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d6866b2006e96d0dfde7de401d3df9962cad60c8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-21 20:10:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"838b9bdb9905a29b6221f719bb8ba86cdf0d7d69","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-21 20:13:46.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"cefed6fbac2e96544139a52f8ead8c6e48ac1693","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 20:29:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8317f2c6cd31485b86c79af71bfa51e1 : SUCCESS in 7m 47s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ce95cc4127cf4436a29ffbdca1524c2b : SUCCESS in 4m 56s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/5af629e003344e15af39a8c2dc0d5525 : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":2},{"id":"13895130e9b4093ff864effd514323f1ee0f3fba","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2020-04-22 09:24:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"10be63433d07ccaa3b4169b27e71ab4b0cbdd06f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-22 15:28:34.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"4dc44ed1a2da3351864033119567373b5607f9c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 15:54:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4ab051fd5887489481c4648eb2303fcc : SUCCESS in 9m 35s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7d76f0e23836403ebe68338592de7755 : SUCCESS in 5m 18s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/64ad030581fe4580b0e5a18537899f36 : SUCCESS in 9m 51s","accounts_in_message":[],"_revision_number":3},{"id":"1ac488b462d759aea20a1944ca54b952bbf2861f","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-04-24 03:06:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5ab3f214c571a0c5983fe32dbdaf3ef7525294a8","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-04-24 13:28:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2d72d4091bc17d6c8b1c89f7c953d81e6255a795","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 13:29:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"823a1e366e25671ff5d89b8f47de9f332683de0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 13:43:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/47c668c4d9f04a3da305fd6af86fbf89 : SUCCESS in 7m 57s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/41860bfe700f4fd78a23c84f0f8bfd15 : SUCCESS in 8m 05s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/88549613b63b4303a6ed57d6b3109824 : SUCCESS in 9m 29s","accounts_in_message":[],"_revision_number":3},{"id":"0fd324416364090900f4fd133c0a083edd9c016d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 13:43:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"a969ffb53cd59ae76178310aee5c850376af2434","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 14:08:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/60707b8bad734bb09b6c926c341973bc : SUCCESS in 3m 42s\n- infra-prod-remote-puppet-else https://zuul.opendev.org/t/openstack/build/9ff8c331346f4a2f8f6ac8eab52c4546 : SUCCESS in 12m 08s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2cdd7f126ed05f9242330551d1dd93150f41b00d","revisions":{"c224ebc8ea8cb61813f87fb78ad51825d9f0da2d":{"kind":"REWORK","_number":1,"created":"2020-04-21 20:10:20.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/22/721722/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/721722/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/721722/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/721722/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/721722/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/721722/1"}}},"commit":{"parents":[{"commit":"9bb8e19da31da6a31109fa1829fef07cae6c32f5","subject":"Retire i18n-specs repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9bb8e19da31da6a31109fa1829fef07cae6c32f5"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-21 20:09:37.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-21 20:09:37.000000000","tz":120},"subject":"Finish retiring i18n-specs","message":"Finish retiring i18n-specs\n\nRemove repo from infrastructure, it\u0027s empty now.\n\nDepends-On: https://review.opendev.org/721720\nChange-Id: Iec11018b3a17c7667f91e27856e68e922cae8336\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c224ebc8ea8cb61813f87fb78ad51825d9f0da2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c224ebc8ea8cb61813f87fb78ad51825d9f0da2d"}]},"branch":"refs/heads/master"},"73f9d84d37fec8793be45b32ba6fbcc17d96e0bb":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-04-21 20:13:46.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/22/721722/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/721722/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/721722/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/721722/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/721722/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/721722/2"}}},"commit":{"parents":[{"commit":"9bb8e19da31da6a31109fa1829fef07cae6c32f5","subject":"Retire i18n-specs repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9bb8e19da31da6a31109fa1829fef07cae6c32f5"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-21 20:09:37.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-21 20:13:01.000000000","tz":120},"subject":"Finish retiring i18n-specs","message":"Finish retiring i18n-specs\n\nRemove repo from infrastructure, it\u0027s empty now.\n\nDepends-On: https://review.opendev.org/721720\nNeeded-By: https://review.opendev.org/721723\nChange-Id: Iec11018b3a17c7667f91e27856e68e922cae8336\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/73f9d84d37fec8793be45b32ba6fbcc17d96e0bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/73f9d84d37fec8793be45b32ba6fbcc17d96e0bb"}]},"branch":"refs/heads/master"},"2cdd7f126ed05f9242330551d1dd93150f41b00d":{"kind":"REWORK","_number":3,"created":"2020-04-22 15:28:34.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/22/721722/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/22/721722/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/721722/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/721722/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/22/721722/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/22/721722/3"}}},"commit":{"parents":[{"commit":"85dae51f3844440be252942758cc8b8997a1428c","subject":"Fix i18n-specs jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/85dae51f3844440be252942758cc8b8997a1428c"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-21 20:09:37.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-22 15:28:08.000000000","tz":120},"subject":"Finish retiring i18n-specs","message":"Finish retiring i18n-specs\n\nRemove repo from infrastructure, it\u0027s empty now.\n\nDepends-On: https://review.opendev.org/721720\nNeeded-By: https://review.opendev.org/721723\nChange-Id: Iec11018b3a17c7667f91e27856e68e922cae8336\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2cdd7f126ed05f9242330551d1dd93150f41b00d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2cdd7f126ed05f9242330551d1dd93150f41b00d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
