)]}'
{"id":"openstack%2Fproject-config~721978","triplet_id":"openstack%2Fproject-config~master~Ia1ac4727a47350a33d5b8d8986664d887f935fde","project":"openstack/project-config","branch":"master","topic":"deprecateAIAB","hashtags":[],"change_id":"Ia1ac4727a47350a33d5b8d8986664d887f935fde","subject":"Retire airship-in-a-bottle: end project gating","status":"MERGED","created":"2020-04-22 13:57:20.000000000","updated":"2020-04-24 19:04:38.000000000","submitted":"2020-04-24 19:04:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721978-1587755078900-354c625b","meta_rev_id":"e97bde445184559becbdc9daeb652cd1a9208606","_number":721978,"virtual_id_number":721978,"owner":{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-24 17:05:26.000000000","_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-24 19:04:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},{"value":2,"date":"2020-04-24 18:52:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-04-24 18:34:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},{"value":1,"date":"2020-04-24 18:52:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-24 18:34:22.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-24 18:52:07.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-04-24 19:04:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e35bf440630566603ba71a17d38b1c3ffc622ca","author":{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},"date":"2020-04-22 13:57:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d26f35ba295d2feea31c7ca2fbec604417fe875","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 14:07:39.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Configuration item has more than one key.  Each zuul.yaml\n  configuration file must be a list of dictionaries with a\n  single key, for example:\n  \n  - job:\n      name: foo\n  \n  - project:\n      name: bar\n  \n  Ensure that every item in the list is a dictionary with only\n  one key (in this example, \u0027job\u0027 and \u0027project\u0027).  This error\n  may be caused by insufficient indentation of the keys under\n  the configuration item (\u0027name\u0027 in this example).","accounts_in_message":[],"_revision_number":1},{"id":"9d74caa997a7516cf68007e3fa247973fdbf2406","author":{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},"date":"2020-04-22 14:48:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bc47c5ed6d4f490e07202f426c89c692fc5c80ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 15:02:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/91ee86db5c46419e82bc79603a6340e7 : SUCCESS in 8m 53s","accounts_in_message":[],"_revision_number":2},{"id":"8873989c13c027f539903f6d9fd1821dfd7a16d5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-22 15:46:15.000000000","message":"Patch Set 2:\n\nPlease fix https://review.opendev.org/#/c/720160/ first\n\nOnce that one is complete, I\u0027ll +2 this one","accounts_in_message":[],"_revision_number":2},{"id":"e625c76e8019277e20ebb0800fb2cd17dfda1b04","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-23 19:10:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e7a97a251fe65744bb9cfa479c2f7c5484beb32d","author":{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},"date":"2020-04-24 17:05:26.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThank you.","accounts_in_message":[],"_revision_number":2},{"id":"699aff25d8733fa866d02a641c4d6cbcb6e5e468","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-24 18:34:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"216d09b4f6cc3c088a674bae49f001d82aab356c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-04-24 18:52:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dd6ca8c021819570a8cb1e829db1cd90fe0e9065","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 18:52:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"35b368d6fa087104558a7dc4e946830c4665f186","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 19:04:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f36abda59837ba70bc3283fb0b9e1d91763e7f90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 19:04:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e98e6b93bff548aea36521e902f66333 : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a37c70a2db47ff37c8df51b87b09d2dd2487f8c2","revisions":{"3ae9c57f29bad3fa9f0d374d04fb359081cd995c":{"kind":"REWORK","_number":1,"created":"2020-04-22 13:57:20.000000000","uploader":{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},"ref":"refs/changes/78/721978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/721978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/721978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/721978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/721978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/721978/1"}}},"commit":{"parents":[{"commit":"cf331b02adb71049058c9ece19438f014581354e","subject":"Merge \"Stop running jobs on unused puppet repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cf331b02adb71049058c9ece19438f014581354e"}]}],"author":{"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","date":"2020-04-22 13:55:56.000000000","tz":120},"committer":{"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","date":"2020-04-22 13:56:47.000000000","tz":120},"subject":"Retire airship-in-a-bottle: end project gating","message":"Retire airship-in-a-bottle: end project gating\n\nNeeded-By: Ic16d19eb805589ab1dbc02cafb8f2d3724ab0302\n\nChange-Id: Ia1ac4727a47350a33d5b8d8986664d887f935fde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3ae9c57f29bad3fa9f0d374d04fb359081cd995c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3ae9c57f29bad3fa9f0d374d04fb359081cd995c"}]},"branch":"refs/heads/master"},"a37c70a2db47ff37c8df51b87b09d2dd2487f8c2":{"kind":"REWORK","_number":2,"created":"2020-04-22 14:48:07.000000000","uploader":{"_account_id":28208,"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","username":"gorshunovr"},"ref":"refs/changes/78/721978/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/721978/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/721978/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/721978/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/721978/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/721978/2"}}},"commit":{"parents":[{"commit":"cf331b02adb71049058c9ece19438f014581354e","subject":"Merge \"Stop running jobs on unused puppet repos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cf331b02adb71049058c9ece19438f014581354e"}]}],"author":{"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","date":"2020-04-22 13:55:56.000000000","tz":120},"committer":{"name":"Roman Gorshunov","email":"roman.gorshunov@att.com","date":"2020-04-22 14:47:47.000000000","tz":120},"subject":"Retire airship-in-a-bottle: end project gating","message":"Retire airship-in-a-bottle: end project gating\n\nNeeded-By: Ic16d19eb805589ab1dbc02cafb8f2d3724ab0302\n\nChange-Id: Ia1ac4727a47350a33d5b8d8986664d887f935fde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a37c70a2db47ff37c8df51b87b09d2dd2487f8c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a37c70a2db47ff37c8df51b87b09d2dd2487f8c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
