)]}'
{"id":"openstack%2Fproject-config~724310","triplet_id":"openstack%2Fproject-config~master~Ia72b3271a65d38ea6b5e4ad03af5813193c47926","project":"openstack/project-config","branch":"master","topic":"add-job","hashtags":[],"change_id":"Ia72b3271a65d38ea6b5e4ad03af5813193c47926","subject":"Add Github mirroring job to all official repos","status":"MERGED","created":"2020-04-29 12:33:47.000000000","updated":"2020-04-29 15:58:09.000000000","submitted":"2020-04-29 15:58:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724310-1588175889811-276f548f","meta_rev_id":"cd8804dd9c8eac7d3ac13a900f793b9e52fb0121","_number":724310,"virtual_id_number":724310,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-29 15:58:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2020-04-29 15:30:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2020-04-29 14:18:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2020-04-29 15:30:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":1,"date":"2020-04-29 15:43:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-29 15:30:39.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2020-04-29 15:43:55.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-29 15:58:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2ee59bbe60128560a9ec2fb8e28c57202e3be639","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-29 12:33:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b120ed372b930119593620481436b95a01c5ca54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 12:35:52.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  The key \"post\" appears more than once; duplicate keys are not\n  permitted.\n\nThe error appears in the following stanza:\n\n  name: openstack/bugdaystats\n      templates:\n        - publish-to-pypi\n      check:\n        jobs:\n          - openstack-tox-pep8\n      gate:\n        jobs:\n          - openstack-tox-pep8\n      post:\n        jobs:\n          - publish-openstack-python-branch-tarball\n      post:\n        jobs:\n          - openstack-upload-github-mirror\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 1510, column 5","accounts_in_message":[],"_revision_number":1},{"id":"cbd6d0036fadaccd8c73d73bb3dd09720813cef3","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-29 12:50:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f3a834a54e683aa765744e365ea17fa385c7e571","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 12:51:10.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  The key \"post\" appears more than once; duplicate keys are not\n  permitted.\n\nThe error appears in the following stanza:\n\n  name: openstack/nose-html-output\n      templates:\n        - noop-jobs\n        - publish-to-pypi\n      post:\n        jobs:\n          - publish-openstack-python-branch-tarball\n      post:\n        jobs:\n          - openstack-upload-github-mirror\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 4085, column 5","accounts_in_message":[],"_revision_number":2},{"id":"7e866cb02d57bdc6702fef506a739dfa371984db","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-29 12:54:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0333401a238fffd4bf5b6cde6e94f5d6433de1c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 12:54:50.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  The key \"post\" appears more than once; duplicate keys are not\n  permitted.\n\nThe error appears in the following stanza:\n\n  name: openstack/openstack-ansible-tests\n      post:\n        jobs:\n          - propose-updates:\n              vars:\n                update_target: openstack-ansible-tests\n      post:\n        jobs:\n          - openstack-upload-github-mirror\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 4617, column 5","accounts_in_message":[],"_revision_number":3},{"id":"76d9555e65c162c5d7243f5abaa93e8a02b1c07d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-29 13:00:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"22760e4b86ea14ddd35b839fece1ad14c3b0c185","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 13:01:19.000000000","message":"Patch Set 4: Verified-1\n\n(6 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  The project \"openstack/i18n-specs\" was not found.  All projects\n  referenced within a Zuul configuration must first be added to the main\n  configuration file by the Zuul administrator.\n\nThe error appears in the following project stanza:\n\n  project:\n      name: openstack/i18n-specs\n      post:\n        jobs:\n          - openstack-upload-github-mirror\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 3016, column 3","accounts_in_message":[],"_revision_number":4},{"id":"958feaf2ded53d4a120d2fb8a153dbc3508a4170","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-29 13:10:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"34bfb6fd1608d2b9ce05222a8fd33c953806ff2b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-29 13:11:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"971cead1783e3cf957a45478f22aea7be89f3eb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 13:18:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4a62406cc0584af89af2dd14e931dc78 : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":5},{"id":"0820adec2c1810267cd290f79a702c05f4f7d278","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-29 14:06:39.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0bdda2ca9d84b84db7d665e06f9eabbf3efb7f63","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-29 14:08:34.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"09762c1bea5ae885bbd0d3a53081b0114961623c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-29 14:12:40.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"c0f2d405d91de648c78a5cd44ccae15bd7b9791c","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-04-29 14:16:38.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"38b82fce4c6d9fad417a35f5be1e70c8439a1638","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-29 14:18:57.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"ea802f4cb6d358c11ee37d4d92426c2a4c53abdc","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-29 14:34:34.000000000","message":"Patch Set 8:\n\nrecheck - event lost","accounts_in_message":[],"_revision_number":8},{"id":"c058125cdfd5e4b6393671fda43c65e67d2c8655","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 14:55:39.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/78ea26c5fcd84bb6ad3a0960b39755c1 : SUCCESS in 9m 39s","accounts_in_message":[],"_revision_number":8},{"id":"5414aa2e5c64632263ea968fa207bbde72728498","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-04-29 15:30:34.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"de9eaf1c6e0563e55876a574aab0e62a45c0ea7e","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-04-29 15:30:39.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"d335b3a6cf1094d86a8eac1edde1df080f1f8d52","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-29 15:43:55.000000000","message":"Patch Set 8: Workflow+1\n\ndependency merged","accounts_in_message":[],"_revision_number":8},{"id":"08469e78cc42a1d8b77d4ff47a276f09350c8619","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 15:45:44.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"3552bd636d324a4aa7958f6cfe66fc9655b9cc71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 15:58:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"05326d51498cd7dcb79cfd08b1b5f5ea62292b6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 15:58:09.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/52fefd6ee74844dfa12589380d3f498c : SUCCESS in 9m 22s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"d8462f4051d7e2dfbb997ae64387896c60cb6f3b","revisions":{"a6536587b2a5f7019f2e5ac2428da30a7d9845f6":{"kind":"REWORK","_number":1,"created":"2020-04-29 12:33:47.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/10/724310/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/724310/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/724310/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/724310/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/724310/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/724310/1"}}},"commit":{"parents":[{"commit":"c30ec5c689d086af875a2a34b000a678c7d5f05c","subject":"Merge \"Normalize projects.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c30ec5c689d086af875a2a34b000a678c7d5f05c"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-29 12:29:19.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-29 12:32:40.000000000","tz":120},"subject":"Add Github mirroring job to all official repos","message":"Add Github mirroring job to all official repos\n\nNow that the Github repositories have been cleaned up of refs/changes\nentries, we can enable per-repo mirroring globally.\n\nNote that:\n- openstack/release-test was already added as a test\n- for repos that did not have an entry in projects.yaml, we create one\n\nNeeded-By: https://review.opendev.org/718478\nChange-Id: Ia72b3271a65d38ea6b5e4ad03af5813193c47926\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a6536587b2a5f7019f2e5ac2428da30a7d9845f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a6536587b2a5f7019f2e5ac2428da30a7d9845f6"}]},"branch":"refs/heads/master"},"171a62e82a3847eab201d012f0861f4ea2e48917":{"kind":"REWORK","_number":2,"created":"2020-04-29 12:50:42.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/10/724310/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/724310/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/724310/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/724310/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/724310/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/724310/2"}}},"commit":{"parents":[{"commit":"c30ec5c689d086af875a2a34b000a678c7d5f05c","subject":"Merge \"Normalize projects.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c30ec5c689d086af875a2a34b000a678c7d5f05c"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-29 12:29:19.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-29 12:50:03.000000000","tz":120},"subject":"Add Github mirroring job to all official repos","message":"Add Github mirroring job to all official repos\n\nNow that the Github repositories have been cleaned up of refs/changes\nentries, we can enable per-repo mirroring globally.\n\nNote that:\n- openstack/release-test was already added as a test\n- for repos that did not have an entry in projects.yaml, we create one\n\nNeeded-By: https://review.opendev.org/718478\nChange-Id: Ia72b3271a65d38ea6b5e4ad03af5813193c47926\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/171a62e82a3847eab201d012f0861f4ea2e48917"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/171a62e82a3847eab201d012f0861f4ea2e48917"}]},"branch":"refs/heads/master"},"368d0a048fe04a93afb96469d65b628ab0eadadd":{"kind":"REWORK","_number":3,"created":"2020-04-29 12:54:11.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/10/724310/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/724310/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/724310/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/724310/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/724310/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/724310/3"}}},"commit":{"parents":[{"commit":"c30ec5c689d086af875a2a34b000a678c7d5f05c","subject":"Merge \"Normalize projects.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c30ec5c689d086af875a2a34b000a678c7d5f05c"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-29 12:29:19.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-29 12:53:56.000000000","tz":120},"subject":"Add Github mirroring job to all official repos","message":"Add Github mirroring job to all official repos\n\nNow that the Github repositories have been cleaned up of refs/changes\nentries, we can enable per-repo mirroring globally.\n\nNote that:\n- openstack/release-test was already added as a test\n- for repos that did not have an entry in projects.yaml, we create one\n\nNeeded-By: https://review.opendev.org/718478\nChange-Id: Ia72b3271a65d38ea6b5e4ad03af5813193c47926\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/368d0a048fe04a93afb96469d65b628ab0eadadd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/368d0a048fe04a93afb96469d65b628ab0eadadd"}]},"branch":"refs/heads/master"},"b71430bfc514dcc0aee85727c3376094e9985852":{"kind":"REWORK","_number":4,"created":"2020-04-29 13:00:55.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/10/724310/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/724310/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/724310/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/724310/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/724310/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/724310/4"}}},"commit":{"parents":[{"commit":"c30ec5c689d086af875a2a34b000a678c7d5f05c","subject":"Merge \"Normalize projects.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c30ec5c689d086af875a2a34b000a678c7d5f05c"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-29 12:29:19.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-29 12:58:58.000000000","tz":120},"subject":"Add Github mirroring job to all official repos","message":"Add Github mirroring job to all official repos\n\nNow that the Github repositories have been cleaned up of refs/changes\nentries, we can enable per-repo mirroring globally.\n\nNote that:\n- openstack/release-test was already added as a test\n- for repos that did not have an entry in projects.yaml, we create one\n\nNeeded-By: https://review.opendev.org/718478\nChange-Id: Ia72b3271a65d38ea6b5e4ad03af5813193c47926\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b71430bfc514dcc0aee85727c3376094e9985852"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b71430bfc514dcc0aee85727c3376094e9985852"}]},"branch":"refs/heads/master"},"c8ec0478f981ec2fff00b7bbbcd9cc522aa467ca":{"kind":"REWORK","_number":5,"created":"2020-04-29 13:10:34.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/10/724310/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/724310/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/724310/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/724310/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/724310/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/724310/5"}}},"commit":{"parents":[{"commit":"c30ec5c689d086af875a2a34b000a678c7d5f05c","subject":"Merge \"Normalize projects.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c30ec5c689d086af875a2a34b000a678c7d5f05c"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-29 12:29:19.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-29 13:10:22.000000000","tz":120},"subject":"Add Github mirroring job to all official repos","message":"Add Github mirroring job to all official repos\n\nNow that the Github repositories have been cleaned up of refs/changes\nentries, we can enable per-repo mirroring globally.\n\nNote that:\n- openstack/release-test was already added as a test\n- for repos that did not have an entry in projects.yaml, we create one\n\nNeeded-By: https://review.opendev.org/718478\nChange-Id: Ia72b3271a65d38ea6b5e4ad03af5813193c47926\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c8ec0478f981ec2fff00b7bbbcd9cc522aa467ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c8ec0478f981ec2fff00b7bbbcd9cc522aa467ca"}]},"branch":"refs/heads/master"},"9b309c1df78b122f7e18aff050947c0615622583":{"kind":"REWORK","_number":6,"created":"2020-04-29 14:06:39.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/10/724310/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/724310/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/724310/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/724310/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/724310/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/724310/6"}}},"commit":{"parents":[{"commit":"c30ec5c689d086af875a2a34b000a678c7d5f05c","subject":"Merge \"Normalize projects.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c30ec5c689d086af875a2a34b000a678c7d5f05c"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-29 13:42:50.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-29 13:54:50.000000000","tz":120},"subject":"Add Github mirroring job to all official repos","message":"Add Github mirroring job to all official repos\n\nNow that the Github repositories have been cleaned up of refs/changes\nentries, we can enable per-repo mirroring globally.\n\nNote that:\n- openstack/release-test was already added as a test\n- for repos that did not have an entry in projects.yaml, we create one\n\nNeeded-By: https://review.opendev.org/718478\nDepends-On: https://review.opendev.org/724329\nChange-Id: Ia72b3271a65d38ea6b5e4ad03af5813193c47926\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9b309c1df78b122f7e18aff050947c0615622583"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9b309c1df78b122f7e18aff050947c0615622583"}]},"branch":"refs/heads/master"},"ee682fab4930a5fe3f491362600992d6d6001004":{"kind":"REWORK","_number":7,"created":"2020-04-29 14:08:34.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/10/724310/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/724310/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/724310/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/724310/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/724310/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/724310/7"}}},"commit":{"parents":[{"commit":"59e06614c1a7e53455fa587836595c59aa2a84f7","subject":"Activate more plain nodes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/59e06614c1a7e53455fa587836595c59aa2a84f7"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-29 14:08:02.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-29 14:08:02.000000000","tz":120},"subject":"Add Github mirroring job to all official repos","message":"Add Github mirroring job to all official repos\n\nNow that the Github repositories have been cleaned up of refs/changes\nentries, we can enable per-repo mirroring globally.\n\nNote that:\n- openstack/release-test was already added as a test\n- for repos that did not have an entry in projects.yaml, we create one\n\nNeeded-By: https://review.opendev.org/718478\nDepends-On: https://review.opendev.org/724329\nChange-Id: Ia72b3271a65d38ea6b5e4ad03af5813193c47926\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ee682fab4930a5fe3f491362600992d6d6001004"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ee682fab4930a5fe3f491362600992d6d6001004"}]},"branch":"refs/heads/master"},"d8462f4051d7e2dfbb997ae64387896c60cb6f3b":{"kind":"REWORK","_number":8,"created":"2020-04-29 14:16:38.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/10/724310/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/724310/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/724310/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/724310/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/724310/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/724310/8"}}},"commit":{"parents":[{"commit":"59e06614c1a7e53455fa587836595c59aa2a84f7","subject":"Activate more plain nodes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/59e06614c1a7e53455fa587836595c59aa2a84f7"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-29 14:08:02.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-04-29 14:15:55.000000000","tz":120},"subject":"Add Github mirroring job to all official repos","message":"Add Github mirroring job to all official repos\n\nNow that the Github repositories have been cleaned up of refs/changes\nentries, we can enable per-repo mirroring globally.\n\nNote that:\n- openstack/release-test was already added as a test\n- for repos that did not have an entry in projects.yaml, we create one\n\nNeeded-By: https://review.opendev.org/718478\nDepends-On: https://review.opendev.org/724329\nChange-Id: Ia72b3271a65d38ea6b5e4ad03af5813193c47926\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d8462f4051d7e2dfbb997ae64387896c60cb6f3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d8462f4051d7e2dfbb997ae64387896c60cb6f3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
