)]}'
{"id":"openstack%2Fproject-config~724886","triplet_id":"openstack%2Fproject-config~master~Idfc6007516b3446541e830db0120a0801f17303e","project":"openstack/project-config","branch":"master","topic":"new-projects","hashtags":[],"change_id":"Idfc6007516b3446541e830db0120a0801f17303e","subject":"Create new project for OpenDev Engagement Stats","status":"MERGED","created":"2020-05-01 16:25:11.000000000","updated":"2020-05-05 20:07:45.000000000","submitted":"2020-05-05 19:49:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724886-1588708153623-57c03af8","meta_rev_id":"51cdd82694595a657f223859a3cdca83b75c29b7","_number":724886,"virtual_id_number":724886,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-05 19:49:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-05-01 16:49:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-05 19:24:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-05 19:24:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-01 16:49:17.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-05 19:24:49.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-05-05 19:49:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c704495e115be3f8876ec5446d2b8039bec0254","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-05-01 16:25:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fdb01c36f278d88832c5e002903bf807662d6a64","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-01 16:29:47.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nLGTM in general, let\u0027s split this up until docs are published","accounts_in_message":[],"_revision_number":1},{"id":"b0312d8748eb55cc6e156a47dfb9580c271b91c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 16:36:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0398ae99d25c4c77bcdb939aa98b4eed : SUCCESS in 6m 55s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0cd91edf7ae940458715447444471b35 : SUCCESS in 4m 50s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/c84f1a0e757243e2ba6a24198efa079b : SUCCESS in 3m 50s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/850598d20b0a4e73b2aecb6aadd0dce4 : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":1},{"id":"cb0291c4fbfeb8addcfc44e0e844f6844377984a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-05-01 16:44:34.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"746bd5c43f98c8175ae342012cf271af0a74c964","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-05-01 16:47:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"89ca595e48108dd9d2123a6b608b581e46d62fa8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-01 16:49:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cf3df90e0c2e684e67eed853d646fade156b7e92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 16:57:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e126c7561da74c4a97dc10a34243551b : SUCCESS in 7m 17s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7bd00ec362f5460f9819c97041aaa38c : SUCCESS in 4m 53s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/64808ac0aa664ef5b63ad8e682a57166 : SUCCESS in 7m 47s","accounts_in_message":[],"_revision_number":2},{"id":"dad91bcdd3a22d94da4b50edc34043c308c7799f","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-05 19:24:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2592b705c618f62511b9097a0d0bbc801a85bcd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 19:25:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b2e917f41fee12622a8cdc68a583e5ea533773ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 19:49:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/cbb21eb951f14469a54267cfac973dd8 : SUCCESS in 7m 09s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5eb1fb9994bc4367bfa1f966994efe81 : SUCCESS in 4m 02s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/98f885fbd1e34492b3445be1c705b4a5 : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":2},{"id":"306784ae7baa2d4c2d6de9931dee57d16ae0a190","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 19:49:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b17bec61483208a94607445f132d3f9aa6a9a666","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 20:07:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/bed427a7c134458fa9c65d2c4b4874cb : SUCCESS in 2m 43s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/6b6506c6015d4877bb00e6db8fa8445d : SUCCESS in 10m 29s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"32982452f11ae4c6e67fdb50034cb21b065d73dc","revisions":{"12c9d96620624717ab38bf9f4ed21ef6cadd830b":{"kind":"REWORK","_number":1,"created":"2020-05-01 16:25:11.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/86/724886/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/86/724886/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/724886/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/724886/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/724886/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/86/724886/1"}}},"commit":{"parents":[{"commit":"f3382545b45e55b3467a5fe77e5e9777e916a96b","subject":"Remove zuul-jobs notification from #openstack-infra","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f3382545b45e55b3467a5fe77e5e9777e916a96b"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2020-05-01 16:20:55.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2020-05-01 16:25:01.000000000","tz":0},"subject":"Create new project for OpenDev Engagement Stats","message":"Create new project for OpenDev Engagement Stats\n\nInitialize a fresh repository to hold tooling used for generating\ncoarse-grained reports of aggregate collaboration activity from\npublicly available APIs and archives provided by OpenDev hosted\nservices.\n\nChange-Id: Idfc6007516b3446541e830db0120a0801f17303e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/12c9d96620624717ab38bf9f4ed21ef6cadd830b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/12c9d96620624717ab38bf9f4ed21ef6cadd830b"}]},"branch":"refs/heads/master"},"32982452f11ae4c6e67fdb50034cb21b065d73dc":{"kind":"REWORK","_number":2,"created":"2020-05-01 16:47:37.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/86/724886/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/86/724886/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/724886/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/724886/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/724886/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/86/724886/2"}}},"commit":{"parents":[{"commit":"f3382545b45e55b3467a5fe77e5e9777e916a96b","subject":"Remove zuul-jobs notification from #openstack-infra","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f3382545b45e55b3467a5fe77e5e9777e916a96b"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2020-05-01 16:20:55.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2020-05-01 16:45:33.000000000","tz":0},"subject":"Create new project for OpenDev Engagement Stats","message":"Create new project for OpenDev Engagement Stats\n\nInitialize a fresh repository to hold tooling used for generating\ncoarse-grained reports of aggregate collaboration activity from\npublicly available APIs and archives provided by OpenDev hosted\nservices.\n\nChange-Id: Idfc6007516b3446541e830db0120a0801f17303e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/32982452f11ae4c6e67fdb50034cb21b065d73dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/32982452f11ae4c6e67fdb50034cb21b065d73dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
