)]}'
{"id":"openstack%2Fproject-config~725751","triplet_id":"openstack%2Fproject-config~master~I36efdede5c3feccb90e470590da146c0be406814","project":"openstack/project-config","branch":"master","topic":"suse-python3-interp","hashtags":[],"change_id":"I36efdede5c3feccb90e470590da146c0be406814","subject":"Stop building opensuse-15-plain images","status":"MERGED","created":"2020-05-06 02:45:48.000000000","updated":"2020-05-08 01:22:45.000000000","submitted":"2020-05-07 20:06:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725751-1588881988001-2682b9c6","meta_rev_id":"cab628cdc22a5c1a4100318895ef9433088d7eee","_number":725751,"virtual_id_number":725751,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-07 20:06:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-05-07 19:54:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-05-07 19:20:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2020-05-07 19:54:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-07 19:20:30.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-07 19:54:07.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-05-07 20:06:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9f479adaf9578815c1f4b5fb3c8c1451bdce7060","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-06 02:45:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0286d59b48c4d28bb97106a65264e878cf8ecf87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 02:53:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/35ccbcc515364bef8f6a2104a5521c1d : SUCCESS in 7m 09s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/8da9778d0be845eb904b1a9d3ae4d3ea : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":1},{"id":"30cf7c4a82a01d3e1db60e5ee58ca58ef6006ddb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-06 06:01:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9c27f2f690d0ab8cd4bfc7a13ee224a0cff56505","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-07 19:20:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"813a7231852f48cbf4c5c59a04e4ccab4cfb2fdd","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-07 19:20:25.000000000","message":"Patch Set 2:\n\nrebased to resolve merge conflict","accounts_in_message":[],"_revision_number":2},{"id":"8e44be807bf5c049456c0b7f80f62f7e8772abf4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-07 19:20:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2484cdee5896101026845849be350dd121004e62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 19:31:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/77db5d52e42c4fa4aff3c4e2d7d771a0 : SUCCESS in 7m 24s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/06f3b99d26e34a4480ae164aac2580d8 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":2},{"id":"53789fe5fb7af38bae3dd351ef6fc94761322641","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-05-07 19:54:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"64842f1d6ded48b2ccc72197955e4f5d3916ae52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 19:54:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ebdd0b9ab9198187b4cc7c5e867ff7993cb8a8fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 20:06:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bca389509ec24b66b9a82b27106a8b1f : SUCCESS in 8m 00s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/715ffcf8428241cb952b8acac62778c7 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"072c24fcd9108be5c1dde8326c953c584786d3f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 20:06:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4ff467dc70ea1612edb1af9cc63fb0111fd91070","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 01:22:45.000000000","message":"Patch Set 2:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/613be63ac84949b4939fa742856e2b5e : FAILURE in 5m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9e38d47f8ec1971b81bfae9a958174e9b0be4a05","revisions":{"3d17efdad2dbfc68ae1dfc488cc37b06ae27331b":{"kind":"REWORK","_number":1,"created":"2020-05-06 02:45:48.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/51/725751/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/51/725751/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/725751/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/725751/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/725751/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/51/725751/1"}}},"commit":{"parents":[{"commit":"92cf529c148be9c4ce131ef50cccf7312658c043","subject":"Drop pip-and-virtualenv from SUSE 15","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/92cf529c148be9c4ce131ef50cccf7312658c043"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-06 02:43:44.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-06 02:43:44.000000000","tz":600},"subject":"Stop building opensuse-15-plain images","message":"Stop building opensuse-15-plain images\n\nDepends-On: https://review.opendev.org/725750\nChange-Id: I36efdede5c3feccb90e470590da146c0be406814\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3d17efdad2dbfc68ae1dfc488cc37b06ae27331b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3d17efdad2dbfc68ae1dfc488cc37b06ae27331b"}]},"branch":"refs/heads/master"},"9e38d47f8ec1971b81bfae9a958174e9b0be4a05":{"kind":"REWORK","_number":2,"created":"2020-05-07 19:20:05.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/51/725751/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/51/725751/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/725751/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/725751/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/725751/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/51/725751/2"}}},"commit":{"parents":[{"commit":"9178ab94a0b591b27244f4562d55d0bdf655de72","subject":"Merge \"Remove special nb04 config file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9178ab94a0b591b27244f4562d55d0bdf655de72"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-06 02:43:44.000000000","tz":600},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-07 19:18:47.000000000","tz":120},"subject":"Stop building opensuse-15-plain images","message":"Stop building opensuse-15-plain images\n\nDepends-On: https://review.opendev.org/725750\nChange-Id: I36efdede5c3feccb90e470590da146c0be406814\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9e38d47f8ec1971b81bfae9a958174e9b0be4a05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9e38d47f8ec1971b81bfae9a958174e9b0be4a05"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
