)]}'
{"id":"openstack%2Fproject-config~725847","triplet_id":"openstack%2Fproject-config~master~I5b12f0343cbb9348e75f090ae59115035970372c","project":"openstack/project-config","branch":"master","topic":"oslo-metrics","hashtags":[],"change_id":"I5b12f0343cbb9348e75f090ae59115035970372c","subject":"Add repository for oslo.metrics","status":"MERGED","created":"2020-05-06 12:49:02.000000000","updated":"2020-05-07 13:45:38.000000000","submitted":"2020-05-07 13:21:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725847-1588857706067-e143bcec","meta_rev_id":"2a8fdd7b7a4a60a06863adf3c898e1b5ce8aa86a","_number":725847,"virtual_id_number":725847,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-05-06 16:01:09.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-07 13:21:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"recommended":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2020-05-06 16:19:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2020-05-06 16:33:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2020-05-07 13:11:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2020-05-07 13:11:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-06 16:19:42.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-05-06 16:33:14.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-07 13:11:39.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-05-07 13:21:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42f0517828f225afd173c2c320342dfc08a0b883","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-05-06 12:49:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c4ef1526707fc516d9a46ed2614d3c658e89b73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 12:58:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/92cda4825bd14efb89b431e0825df295 : SUCCESS in 7m 44s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/11ee7ef7f93a4cdd89b995dd7863230a : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"b239b5f4ec0704af69816cd410d414cda5b4f996","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-06 14:30:30.000000000","message":"Patch Set 1:\n\nDo you want to push up a followup change to add the github sync template?","accounts_in_message":[],"_revision_number":1},{"id":"b312785ad855c39fee671949b2b1022a7da5f65c","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-05-06 16:01:09.000000000","message":"Patch Set 1:\n\nI was thinking we should wait until it has content to sync it... which should take a week or so","accounts_in_message":[],"_revision_number":1},{"id":"f6ae52907f51d3d87172aa29bcd992053ff10b04","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-06 16:10:22.000000000","message":"Patch Set 1:\n\n\u003e I was thinking we should wait until it has content to sync it...\n \u003e which should take a week or so\n\njust don\u0027t forget it.\n\nChange itself LGTM, waiting for PTL to +1 before I +2.","accounts_in_message":[],"_revision_number":1},{"id":"ec99f72b8c7adb3ab0926d4489c3520f6f20ac3b","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-05-06 16:19:42.000000000","message":"Patch Set 1: Code-Review+1\n\nHappy to see this moving along!","accounts_in_message":[],"_revision_number":1},{"id":"49a03b7dd9bc3cc4d9b2fe7667655618aff5fa46","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-06 16:33:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4a4a9dfa21a5c9141a787fb08a105c6d9aa42812","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-05-07 13:11:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9cf64a65672d107700eb8adf5f78e2a12b989d7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 13:11:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5cdb028c272e1fc81146148465ed4fc26daacea7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 13:21:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f4b96b7ff9f7463a9971fc984ce51441 : SUCCESS in 8m 36s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ae8bd48d81c74617b32ae2d5a7039b87 : SUCCESS in 7m 47s","accounts_in_message":[],"_revision_number":1},{"id":"070c8b43b9b3bf72fb3f210e749ab61f9d562904","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 13:21:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d97feebf7621c053b8736c5b750fb8ba164ec3bc","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 13:45:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/3d5fa7588b124de5824806205d0c3f5e : SUCCESS in 2m 56s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/7ed2821da6c948e4bf877c097c17c385 : SUCCESS in 10m 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2146c3afc1e0221ed0c012268454bb3381c5a727","revisions":{"2146c3afc1e0221ed0c012268454bb3381c5a727":{"kind":"REWORK","_number":1,"created":"2020-05-06 12:49:02.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/47/725847/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/47/725847/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/725847/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/725847/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/725847/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/47/725847/1"}}},"commit":{"parents":[{"commit":"abf5d4af1e173d6d05ec4b0fd4243c3c2458d7d1","subject":"Merge \"Clean up unused IRC channels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/abf5d4af1e173d6d05ec4b0fd4243c3c2458d7d1"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-05-06 12:36:47.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-05-06 12:46:46.000000000","tz":120},"subject":"Add repository for oslo.metrics","message":"Add repository for oslo.metrics\n\nFollowing [1], initial code for oslo.metrics should soon be proposed.\nLet\u0027s create a repository in OpenDev to host that code.\n\n[1] https://specs.openstack.org/openstack/oslo-specs/specs/ussuri/oslo-metrics.html\n\nNeeded-By: I454700e72ba8c9a7a59e69c97011884ea5b7ae15\nChange-Id: I5b12f0343cbb9348e75f090ae59115035970372c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2146c3afc1e0221ed0c012268454bb3381c5a727"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2146c3afc1e0221ed0c012268454bb3381c5a727"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
