)]}'
{"id":"openstack%2Fproject-config~726043","triplet_id":"openstack%2Fproject-config~master~I55cfedb9ac7fa06ba1725f0ec92f7247237dc74b","project":"openstack/project-config","branch":"master","topic":"nb-opendev","hashtags":[],"change_id":"I55cfedb9ac7fa06ba1725f0ec92f7247237dc74b","subject":"Remove special nb04 config file","status":"MERGED","created":"2020-05-07 05:53:44.000000000","updated":"2020-05-07 15:33:51.000000000","submitted":"2020-05-07 15:13:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726043-1588864436687-23512575","meta_rev_id":"94b529509a0ff729308ca4040c7dd575e79d28e4","_number":726043,"virtual_id_number":726043,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-07 15:13:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-05-07 15:04:12.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-05-07 07:40:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2020-05-07 15:04:12.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-07 07:40:34.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-07 15:04:12.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-05-07 15:13:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"86bbed915ba5194368f4688a058a46510950f3c2","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-07 05:53:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"343093e66440bb9d0b76380965810b614f654636","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-07 05:54:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2c10c18d334c19a066618ac395cb49d834653d72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 06:01:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/61e02904dcf049b1849e707e5b16583c : SUCCESS in 6m 29s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/ffa786b7c37c471ebf358efeaf189e5d : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":2},{"id":"5948d180a86d69014d544069e8a68bf103830201","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-07 07:40:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a45e49e366c33a55a24bf875158fcac20cbfe517","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-05-07 15:04:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0b53189a9051b372369c05d0b07e06d2ea774362","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 15:04:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d37478e3d5af562a6e3d3c86c6f4d4267989543c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 15:13:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/49f19362a76e46b7adf6bc06cc5bae85 : SUCCESS in 8m 34s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/3ca3cb356e2a4ebab90618f47b85c409 : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":2},{"id":"00d7003b83e1dc52ee0ad249508f45a093d23301","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 15:13:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f0eca29c3709a39bac932f886545107c0e6ba925","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 15:33:51.000000000","message":"Patch Set 2:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/5ce869e00e044ed1b6f328fc7f8265d7 : FAILURE in 7m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4ac03a4a83a2991e10a259e0f5d2361218b83f13","revisions":{"9cb14af03ef6f221502b8b436d223d28e3934c04":{"kind":"REWORK","_number":1,"created":"2020-05-07 05:53:44.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/43/726043/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/43/726043/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/726043/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/726043/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/726043/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/43/726043/1"}}},"commit":{"parents":[{"commit":"22554bfc2a3376f5637499c81c1a930284e3ea26","subject":"Merge \"Drop pip-and-virtualenv from SUSE 15\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/22554bfc2a3376f5637499c81c1a930284e3ea26"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-07 05:52:09.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-07 05:52:09.000000000","tz":600},"subject":"Remove special nb04 config file","message":"Remove special nb04 config file\n\nWith nb01/02.opendev.org now being Ansible/container builders, all\nbuilders can use the same config.  This moves the nb04 config back to\nnodepool.yaml and unpauses everything except gentoo, which is still\nconstantly failing.\n\nChange-Id: I55cfedb9ac7fa06ba1725f0ec92f7247237dc74b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9cb14af03ef6f221502b8b436d223d28e3934c04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9cb14af03ef6f221502b8b436d223d28e3934c04"}]},"branch":"refs/heads/master"},"4ac03a4a83a2991e10a259e0f5d2361218b83f13":{"kind":"REWORK","_number":2,"created":"2020-05-07 05:54:45.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/43/726043/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/43/726043/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/726043/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/726043/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/726043/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/43/726043/2"}}},"commit":{"parents":[{"commit":"22554bfc2a3376f5637499c81c1a930284e3ea26","subject":"Merge \"Drop pip-and-virtualenv from SUSE 15\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/22554bfc2a3376f5637499c81c1a930284e3ea26"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-07 05:52:09.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-07 05:54:29.000000000","tz":600},"subject":"Remove special nb04 config file","message":"Remove special nb04 config file\n\nWith nb01/02.opendev.org now being Ansible/container builders, all\nbuilders can use the same config.  This moves the nb04 config back to\nnodepool.yaml and unpauses everything except gentoo, which is still\nconstantly failing.\n\nChange-Id: I55cfedb9ac7fa06ba1725f0ec92f7247237dc74b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4ac03a4a83a2991e10a259e0f5d2361218b83f13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4ac03a4a83a2991e10a259e0f5d2361218b83f13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
