)]}'
{"id":"openstack%2Fproject-config~726461","triplet_id":"openstack%2Fproject-config~master~I3fdbcc8f8e0014a67f5ff1fcdd429d12329e0d16","project":"openstack/project-config","branch":"master","topic":"retire-pbrx","hashtags":[],"change_id":"I3fdbcc8f8e0014a67f5ff1fcdd429d12329e0d16","subject":"Retire x/pbrx - part 1","status":"MERGED","created":"2020-05-08 19:16:38.000000000","updated":"2020-05-13 16:11:36.000000000","submitted":"2020-05-13 16:11:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726461-1589386296184-8773121f","meta_rev_id":"4b26d899770e22210e56ab26b5ab3315565682ca","_number":726461,"virtual_id_number":726461,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-13 16:11:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-05-08 19:58:06.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-05-13 15:59:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2020-05-13 15:59:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"date":"2020-05-13 13:08:22.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-08 19:58:06.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-05-13 15:59:00.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-05-13 16:11:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"884669ab62bbe33581295172ccb3aac12612d449","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-08 19:16:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f94939812a5cdac80ebe2f936760809df8cbf0fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 19:17:02.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  The project \"openstack/pbrx\" was not found.  All projects referenced\n  within a Zuul configuration must first be added to the main\n  configuration file by the Zuul administrator.\n\nThe error appears in the following project stanza:\n\n  project:\n      name: openstack/pbrx\n      templates:\n        - noop-jobs\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 8029, column 3","accounts_in_message":[],"_revision_number":1},{"id":"764c9e47ddfdec099c7d8f5c187964edb4369fe1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-08 19:21:22.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"354bb4571a1eb9e2789a3a9882e47fd7cee9f76f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 19:21:49.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  The project \"openstack/pbrx\" was not found.  All projects referenced\n  within a Zuul configuration must first be added to the main\n  configuration file by the Zuul administrator.\n\nThe error appears in the following project stanza:\n\n  project:\n      name: openstack/pbrx\n      templates:\n        - noop-jobs\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 8029, column 3","accounts_in_message":[],"_revision_number":2},{"id":"c96d0524cc474fcf4369d950c25d6bc32ee9f314","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-08 19:32:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d0aa700718885d36d51454e0cb8f41edf8c05455","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 19:44:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f5f2e9b90d6941deae5a8b6610d14a52 : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":3},{"id":"de02c61f5ecd445d2f1201b788a218807b7ebe57","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-05-08 19:58:06.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0bbe4bfadb14c75239830a086de7a2df7b960ce9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-09 05:39:43.000000000","message":"Patch Set 3: Workflow-1\n\nhttps://opendev.org/openstack/openstackclient/src/branch/master/playbooks/openstackclient-check-plugins/run.yaml#L13 will break, we need to change that first","accounts_in_message":[],"_revision_number":3},{"id":"9d597b8abf999338e618ef62705ea8de95a01dda","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-13 13:08:22.000000000","message":"Patch Set 3: -Workflow\n\nopenstackclient fixed","accounts_in_message":[],"_revision_number":3},{"id":"07f3b0222673cddcdb1365c2cbde509b0aa971d8","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-05-13 15:59:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b9f8a1f6e1522a03d75cefaae182f5123e8c7ebb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 15:59:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"dbe36cb48a3108317e2edf0b8c24bb9602bdbdf1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 16:11:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4d0bbc326aa24b9ba54eba1659362881 : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":3},{"id":"b7e04336f92529a3d759f59d22fb220409abbce5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 16:11:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9ccdffeee94de064b16e83f41625f279265eeef5","revisions":{"09a04c2f40ff94c772af380487e06d97af6e2935":{"kind":"REWORK","_number":1,"created":"2020-05-08 19:16:38.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/61/726461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/61/726461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/726461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/726461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/726461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/61/726461/1"}}},"commit":{"parents":[{"commit":"57ba8ae454a2270eaf1214c9ac617b7c977d29c7","subject":"Remove dmsimard from accessbot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/57ba8ae454a2270eaf1214c9ac617b7c977d29c7"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-08 19:07:13.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-08 19:07:13.000000000","tz":120},"subject":"Retire x/pbrx - part 1","message":"Retire x/pbrx - part 1\n\nThis repo is unused and obsolete.\n\nAnnouncement:\nhttp://lists.zuul-ci.org/pipermail/zuul-discuss/2020-May/001233.html\n\nChange-Id: I3fdbcc8f8e0014a67f5ff1fcdd429d12329e0d16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/09a04c2f40ff94c772af380487e06d97af6e2935"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/09a04c2f40ff94c772af380487e06d97af6e2935"}]},"branch":"refs/heads/master"},"5ad4bd148fc68a1ec200233e76b745e2920a3368":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-08 19:21:22.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/61/726461/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/61/726461/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/726461/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/726461/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/726461/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/61/726461/2"}}},"commit":{"parents":[{"commit":"57ba8ae454a2270eaf1214c9ac617b7c977d29c7","subject":"Remove dmsimard from accessbot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/57ba8ae454a2270eaf1214c9ac617b7c977d29c7"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-08 19:07:13.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-08 19:18:43.000000000","tz":120},"subject":"Retire x/pbrx - part 1","message":"Retire x/pbrx - part 1\n\nThis repo is unused and obsolete.\n\nAnnouncement:\nhttp://lists.zuul-ci.org/pipermail/zuul-discuss/2020-May/001233.html\n\nNeeded-By: https://review.opendev.org/726462\nChange-Id: I3fdbcc8f8e0014a67f5ff1fcdd429d12329e0d16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5ad4bd148fc68a1ec200233e76b745e2920a3368"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5ad4bd148fc68a1ec200233e76b745e2920a3368"}]},"branch":"refs/heads/master"},"9ccdffeee94de064b16e83f41625f279265eeef5":{"kind":"REWORK","_number":3,"created":"2020-05-08 19:32:40.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/61/726461/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/61/726461/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/726461/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/726461/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/61/726461/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/61/726461/3"}}},"commit":{"parents":[{"commit":"57ba8ae454a2270eaf1214c9ac617b7c977d29c7","subject":"Remove dmsimard from accessbot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/57ba8ae454a2270eaf1214c9ac617b7c977d29c7"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-08 19:07:13.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-08 19:31:44.000000000","tz":120},"subject":"Retire x/pbrx - part 1","message":"Retire x/pbrx - part 1\n\nThis repo is unused and obsolete.\n\nAnnouncement:\nhttp://lists.zuul-ci.org/pipermail/zuul-discuss/2020-May/001233.html\n\nNeeded-By: https://review.opendev.org/726462\nChange-Id: I3fdbcc8f8e0014a67f5ff1fcdd429d12329e0d16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9ccdffeee94de064b16e83f41625f279265eeef5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9ccdffeee94de064b16e83f41625f279265eeef5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
