)]}'
{"id":"openstack%2Fproject-config~726848","triplet_id":"openstack%2Fproject-config~master~If29ca8a06759a3871a46826caaed1d56dda86603","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"If29ca8a06759a3871a46826caaed1d56dda86603","subject":"Temporarily disable OVH","status":"MERGED","created":"2020-05-11 13:40:20.000000000","updated":"2020-05-11 21:09:37.000000000","submitted":"2020-05-11 14:44:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726848-1589208294494-eaa21e4c","meta_rev_id":"9c65de2a547a55808e643d590c4aeb78974f2f0c","_number":726848,"virtual_id_number":726848,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-11 14:44:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-05-11 13:43:24.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-05-11 13:47:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-11 13:41:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-11 13:41:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-11 13:41:43.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-05-11 13:43:24.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-05-11 13:47:41.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-11 14:44:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3320eb75b6e61b3ce045e9bcbfc7bbd9078a4069","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-05-11 13:40:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fde20dc192c7858b60e51c0e3c6609ad569f5fae","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-05-11 13:41:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8db480eb01fc4294eb90abeb8acd91cb8c53fdcf","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-05-11 13:43:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d8cd85b51858cc14fdffa3df84c3f640dd4e9302","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-11 13:47:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5cc2a163810fd48473df5cb0e58d21663c5fb4a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 14:29:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d7cbfcce478d4427af269e6edef28831 : SUCCESS in 10m 47s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/d6dd2f6e9fdf4f399d87f7d359280f7e : SUCCESS in 8m 15s","accounts_in_message":[],"_revision_number":1},{"id":"cf977b725aa20338a3487ab606a70269951eedf9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 14:30:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"73a8d773c1c84c6f8e3e7b8d7a35cda451c240f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 14:44:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3c3a31c1cb4f47cf8acd33a8f9f23386 : SUCCESS in 11m 06s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/b20a15e5fde3464ca8dcd4e675c89e42 : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":1},{"id":"d33e9cb9dd71f6002cb8f575ee9a3a9f042f0d03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 14:44:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a140df581c2f42f8e5cc8b7a48171bb7b5721fbb","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 14:51:11.000000000","message":"Patch Set 1:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/71eedea05c2b41c9a9b1bdb803de4c25 : FAILURE in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"9c65de2a547a55808e643d590c4aeb78974f2f0c","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 21:09:37.000000000","message":"Patch Set 1:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/87c45fc0de7a4830aa5000ac160848dc : FAILURE in 4m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"37d292ee74c1a733b55e0f488c486d9f7b8a8162","revisions":{"37d292ee74c1a733b55e0f488c486d9f7b8a8162":{"kind":"REWORK","_number":1,"created":"2020-05-11 13:40:20.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/48/726848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/48/726848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/726848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/726848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/48/726848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/48/726848/1"}}},"commit":{"parents":[{"commit":"57ba8ae454a2270eaf1214c9ac617b7c977d29c7","subject":"Remove dmsimard from accessbot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/57ba8ae454a2270eaf1214c9ac617b7c977d29c7"}]}],"author":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2020-05-11 13:36:43.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jeblair@redhat.com","date":"2020-05-11 13:36:43.000000000","tz":-420},"subject":"Temporarily disable OVH","message":"Temporarily disable OVH\n\nWe have lost contact with our mirrors and seem unable to access\nour account.\n\nChange-Id: If29ca8a06759a3871a46826caaed1d56dda86603\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/37d292ee74c1a733b55e0f488c486d9f7b8a8162"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/37d292ee74c1a733b55e0f488c486d9f7b8a8162"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
