)]}'
{"id":"openstack%2Fproject-config~727902","triplet_id":"openstack%2Fproject-config~master~Ic5bf84e7194edbac0be816519a5c339f3349d35f","project":"openstack/project-config","branch":"master","topic":"rm-openstack-mirror","hashtags":[],"change_id":"Ic5bf84e7194edbac0be816519a5c339f3349d35f","subject":"Remove citycloud kna1/lon1/sto2 clouds","status":"MERGED","created":"2020-05-14 01:56:27.000000000","updated":"2020-05-15 12:14:40.000000000","submitted":"2020-05-15 12:05:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727902-1589544350046-0ef704bd","meta_rev_id":"b1d08dc156aaa9e7f4cde417589f202250b370ac","_number":727902,"virtual_id_number":727902,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-15 12:05:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"disliked":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":-1,"date":"2020-05-15 03:21:16.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-05-15 11:58:18.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-05-15 03:21:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-05-15 06:08:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2020-05-15 11:58:18.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-14 07:36:04.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-14 20:35:29.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-05-15 06:08:56.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-05-15 11:58:18.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-05-15 12:05:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e6f347f79fe55c95bfb22ea137a2906a12a24a5","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-14 01:56:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea50a6f312a6fc582c5d8919eedb6e08fa110e6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 02:07:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2e7740517b17475587407d577cf2975a : SUCCESS in 6m 32s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/4dab5f07caf6460787ec694fb3094052 : FAILURE in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"bca8c423972261770e2bab539d4f713d77e24a19","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-14 07:27:38.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d0944bd4822fb7f70b0a159c76deb092bf4d93e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-14 07:34:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c7aa18a1d97bca3714e9f9afce888e588db66c32","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-14 07:36:04.000000000","message":"Patch Set 2: Code-Review+2\n\nFixed the yaml","accounts_in_message":[],"_revision_number":2},{"id":"a8901558848ca0a620673127eb01be8c0d09e0b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 07:43:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/55646a2ff51a4283a945f86e82838e2d : SUCCESS in 7m 41s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/e839a009e107463bba508ead0db05194 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":2},{"id":"38dc5a854db28132ea9c578173ac79dbf33fa819","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-05-14 09:13:56.000000000","message":"Patch Set 2:\n\nShouldn\u0027t this be removed from nb04 at the same time?","accounts_in_message":[],"_revision_number":2},{"id":"af7e0de2b31b73527a08ff9cbe76a01126c75beb","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-05-14 20:35:29.000000000","message":"Patch Set 2: Code-Review-1 Workflow+1\n\n\u003e Shouldn\u0027t this be removed from nb04 at the same time?\n\nYes, please. I think we might even be able to cleanup nb04 at this point too? Not sure if we plan to keep a third x86 builder around.","accounts_in_message":[],"_revision_number":2},{"id":"3d0c62f5d2fb834cd2223c48e92950a5095e5393","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 20:35:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"50c18348be686762da94ae7b4d527c3656eed4e3","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-05-14 20:36:26.000000000","message":"Patch Set 2: -Workflow\n\nI didn\u0027t mean to -1 and approve it :/ not sure what happened there.","accounts_in_message":[],"_revision_number":2},{"id":"c7661888a11137fcb6c08aa937a412e707373c84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 20:45:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8e822b0b4fe949bf873c70419b79cb36 : SUCCESS in 9m 03s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/affd706dc8954f48aad9591af9c58056 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":2},{"id":"e2c8b845928d5080908b60438f0ec1f97da6da23","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-15 02:09:26.000000000","message":"Patch Set 2:\n\n\u003e Shouldn\u0027t this be removed from nb04 at the same time?\n\nAll the builders are container based now, so they all use the same config file?  I don\u0027t think there\u0027s anything to do special for nb04.  i agree we can remove it, if we want to just have two builders again.","accounts_in_message":[],"_revision_number":2},{"id":"97e51165586a9dbfa6ac60bcd936876c42cb87b1","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-15 03:21:16.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"1df38e22d124e1882a2471db2ef870853e89ebd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 03:30:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0dbdeb9b04d849b4b4221e35498a47eb : SUCCESS in 6m 58s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/773765e07e5a4fedbf8026e0a430c414 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":3},{"id":"fa5cc2f18526cb5195e76e39695fd6dbec0df9d9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-05-15 06:08:56.000000000","message":"Patch Set 3: Code-Review+2\n\nMy bad, I was looking at an older checkout of the repo where the nb04 file was still in place.","accounts_in_message":[],"_revision_number":3},{"id":"63754f37a014fd3716b89fe803f8ebcd88d5dc6c","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-05-15 11:58:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nI agree with frickler","accounts_in_message":[],"_revision_number":3},{"id":"fb72f9d534cd610d4b0bd34b1610ff6f950c4a30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 11:58:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2fe24e31a3b535d7194a04479d8b3329ed41869d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 12:05:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bd54699d7ad54a55a3275cece4b052eb : SUCCESS in 6m 29s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/b92982f4c5fc4d6ba614890505ac95c0 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":3},{"id":"88b9d60c0ccb6bb379b0b422fbdb8064440b41ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 12:05:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0c4ebe9a0314070a13f119ffaf3fd9117fdfebaa","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 12:14:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/717f36490f11412081f2d4bc39228568 : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"24fa46d314b10df4a7248c27a91f2e9aa2f9f7a6","revisions":{"ab25cecdf553f458af00486f6f212f920af77420":{"kind":"REWORK","_number":1,"created":"2020-05-14 01:56:27.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/02/727902/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/02/727902/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/727902/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/727902/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/727902/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/02/727902/1"}}},"commit":{"parents":[{"commit":"a2ce3916e10c0a598ccfee7101f098f766c5c1db","subject":"Switch nodepool builders to opendev.org mirror","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a2ce3916e10c0a598ccfee7101f098f766c5c1db"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-14 01:55:25.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-14 01:55:25.000000000","tz":600},"subject":"Remove citycloud kna1/lon1/sto2 clouds","message":"Remove citycloud kna1/lon1/sto2 clouds\n\nThese clouds have been disabled for some time, remove them from\nnodepool configuration.\n\nChange-Id: Ic5bf84e7194edbac0be816519a5c339f3349d35f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ab25cecdf553f458af00486f6f212f920af77420"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ab25cecdf553f458af00486f6f212f920af77420"}]},"branch":"refs/heads/master"},"b19da5d3dcbd4882b7283ef2c9babc3f64a9fcb4":{"kind":"REWORK","_number":2,"created":"2020-05-14 07:34:59.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/02/727902/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/02/727902/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/727902/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/727902/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/727902/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/02/727902/2"}}},"commit":{"parents":[{"commit":"a2ce3916e10c0a598ccfee7101f098f766c5c1db","subject":"Switch nodepool builders to opendev.org mirror","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a2ce3916e10c0a598ccfee7101f098f766c5c1db"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-14 01:55:25.000000000","tz":600},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-14 07:28:55.000000000","tz":120},"subject":"Remove citycloud kna1/lon1/sto2 clouds","message":"Remove citycloud kna1/lon1/sto2 clouds\n\nThese clouds have been disabled for some time, remove them from\nnodepool configuration.\n\nChange-Id: Ic5bf84e7194edbac0be816519a5c339f3349d35f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b19da5d3dcbd4882b7283ef2c9babc3f64a9fcb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b19da5d3dcbd4882b7283ef2c9babc3f64a9fcb4"}]},"branch":"refs/heads/master"},"24fa46d314b10df4a7248c27a91f2e9aa2f9f7a6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-05-15 03:21:16.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/02/727902/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/02/727902/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/727902/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/727902/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/727902/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/02/727902/3"}}},"commit":{"parents":[{"commit":"e1b0f2dddf49426580dc4cb523037c41993596c9","subject":"Switch RAX IAD mirror to opendev.org version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e1b0f2dddf49426580dc4cb523037c41993596c9"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-14 01:55:25.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-15 03:16:59.000000000","tz":600},"subject":"Remove citycloud kna1/lon1/sto2 clouds","message":"Remove citycloud kna1/lon1/sto2 clouds\n\nThese clouds have been disabled for some time, remove them from\nnodepool configuration.\n\nChange-Id: Ic5bf84e7194edbac0be816519a5c339f3349d35f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/24fa46d314b10df4a7248c27a91f2e9aa2f9f7a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/24fa46d314b10df4a7248c27a91f2e9aa2f9f7a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
