)]}'
{"id":"openstack%2Fproject-config~727956","triplet_id":"openstack%2Fproject-config~master~I1d4c64a91e07f7e5fd27399720e03ac567658aab","project":"openstack/project-config","branch":"master","topic":"rm-openstack-mirror","hashtags":[],"change_id":"I1d4c64a91e07f7e5fd27399720e03ac567658aab","subject":"Remove Citycloud from grafana","status":"MERGED","created":"2020-05-14 07:34:59.000000000","updated":"2020-05-15 12:07:01.000000000","submitted":"2020-05-15 12:07:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727956-1589544421687-be6e565f","meta_rev_id":"2951a1a8581983820fc4bc3d8f71694dd2d7999f","_number":727956,"virtual_id_number":727956,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-15 12:07:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-05-15 11:56:56.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-05-15 03:21:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2020-05-15 11:56:56.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-15 03:21:16.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-14 09:12:03.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-05-15 03:21:16.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-05-15 11:56:56.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-05-15 12:07:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c7652c0b80aef0fd96a8100cb15b0b8a5a0953f2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-14 07:34:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77aa6a39da1e9bb45af324740282a2eba8583f6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 07:46:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2156797d498f44c59e3f23cc370dae09 : SUCCESS in 9m 19s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/c3d4e11e3eff428d90c087b885f2d984 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"fe534f8bfffd2aad9e35ce5e685dccad8a188c69","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-05-14 09:12:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d8bf0c64e67be96df01dc35ec267d45f289be32e","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-15 03:21:16.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"b515f8dad10a455ab7074f717b1d6b13e307bd26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 03:33:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9efdea98bbf74cef8176c5c787dad325 : SUCCESS in 6m 59s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/2f2c1f723f8e4269937a80cb0d7cfd14 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":2},{"id":"bda29bb76c45958078e238d3fedf2ed5dab3f317","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-05-15 11:56:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2f6bc22b4a1738c7fbf1d1f4f97347d39fa716bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 11:58:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c654615180230d4c9ee58e634f0353c91f06e2a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 12:07:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/099528ee589040ea90b9955c76ed4fe2 : SUCCESS in 8m 07s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/26ec525df3b34aacaecacfc9800d39cd : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":2},{"id":"19ea24141f0cc91abdcb210a84a10bd629a24bae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 12:07:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c972ce4ee4fc061d6362fa3d15e28d820ebd8790","revisions":{"7b47d5379747f28774369362494b656ed865a2c9":{"kind":"REWORK","_number":1,"created":"2020-05-14 07:34:59.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/56/727956/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/56/727956/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/727956/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/727956/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/727956/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/56/727956/1"}}},"commit":{"parents":[{"commit":"b19da5d3dcbd4882b7283ef2c9babc3f64a9fcb4","subject":"Remove citycloud kna1/lon1/sto2 clouds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b19da5d3dcbd4882b7283ef2c9babc3f64a9fcb4"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-14 07:33:52.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-14 07:33:52.000000000","tz":120},"subject":"Remove Citycloud from grafana","message":"Remove Citycloud from grafana\n\nThe Citycloud clouds (except the special airship one) are not used\nanymore, remove them from grafana.\n\nChange-Id: I1d4c64a91e07f7e5fd27399720e03ac567658aab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7b47d5379747f28774369362494b656ed865a2c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7b47d5379747f28774369362494b656ed865a2c9"}]},"branch":"refs/heads/master"},"c972ce4ee4fc061d6362fa3d15e28d820ebd8790":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-15 03:21:16.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/56/727956/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/56/727956/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/727956/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/727956/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/727956/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/56/727956/2"}}},"commit":{"parents":[{"commit":"24fa46d314b10df4a7248c27a91f2e9aa2f9f7a6","subject":"Remove citycloud kna1/lon1/sto2 clouds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/24fa46d314b10df4a7248c27a91f2e9aa2f9f7a6"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-14 07:33:52.000000000","tz":120},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-15 03:17:19.000000000","tz":600},"subject":"Remove Citycloud from grafana","message":"Remove Citycloud from grafana\n\nThe Citycloud clouds (except the special airship one) are not used\nanymore, remove them from grafana.\n\nChange-Id: I1d4c64a91e07f7e5fd27399720e03ac567658aab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c972ce4ee4fc061d6362fa3d15e28d820ebd8790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c972ce4ee4fc061d6362fa3d15e28d820ebd8790"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
