)]}'
{"id":"openstack%2Fproject-config~728332","triplet_id":"openstack%2Fproject-config~master~Ic19ffc6dda130ecba2e70bed6bac9fe7a8b8cba9","project":"openstack/project-config","branch":"master","topic":"rm-openstack-mirror","hashtags":[],"change_id":"Ic19ffc6dda130ecba2e70bed6bac9fe7a8b8cba9","subject":"Remove linaro-london cloud","status":"MERGED","created":"2020-05-15 02:17:08.000000000","updated":"2020-05-15 12:22:09.000000000","submitted":"2020-05-15 12:07:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728332-1589544423691-299f2a5a","meta_rev_id":"60aa3aa2ddd1006a93f85d6d24bae6e6afd268cb","_number":728332,"virtual_id_number":728332,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-15 12:07:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-05-15 11:56:21.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-05-15 07:10:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-05-15 03:21:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2020-05-15 11:56:21.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-15 02:21:32.000000000","updated_by":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"reviewer":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"state":"REVIEWER"},{"updated":"2020-05-15 07:10:51.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-15 11:56:21.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-05-15 12:07:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"82fa0503bf9efdb4f24e75c34c57499733b429b8","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-15 02:17:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d17582d195e455641d5ed5720a74d5bc2d9a9bc","author":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"date":"2020-05-15 02:21:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4828fb6d0a844c860a815ef82aaa96497cbf7026","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 02:25:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/85a3b0931bce4016bd0e0c56bf424b0b : SUCCESS in 7m 29s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/3d19d205449a49d49338f71ba605c825 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"567f02d99e3799f6cebb239172fbdf4acae054f1","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-15 03:21:16.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"37e0dee68832a60a061746c032c5c6399912c515","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 03:33:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f1ad9b9f74e14d04b813143dcb5c8181 : SUCCESS in 7m 17s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/83fde5182cdc4ff381063f2dcefe7ce6 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":2},{"id":"41e431b7451ce4cd06dfa016abcf13121ba3587c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-15 07:10:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9f7ce189e370a8e6ae0b7e79a8ae45d986e3f192","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-05-15 11:56:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d662f12f790904ed242e03723dba9a01bf90902b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 11:58:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a70e8be3039e13fc8ab4f6fc93c25900fbac6e0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 12:07:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6d9cb27a921dc17e9438ab00fe593b6c678cca96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 12:07:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fefe1651c63f4508b22315fbcfcd45a5 : SUCCESS in 7m 35s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/22a079755c5f4f4e9c7b9c0fa59192e3 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":2},{"id":"f7dce980bd40b3c76bcd2753a830465b3236657f","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 12:22:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/e215878cd7634deea034283df96aefe1 : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1bb58d216d7b8560af5abf49994abb3be304a27e","revisions":{"6b332b0d2300e558c8bc1bbf928eaa552489b38f":{"kind":"REWORK","_number":1,"created":"2020-05-15 02:17:08.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/32/728332/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/32/728332/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/728332/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/728332/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/728332/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/32/728332/1"}}},"commit":{"parents":[{"commit":"7b47d5379747f28774369362494b656ed865a2c9","subject":"Remove Citycloud from grafana","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7b47d5379747f28774369362494b656ed865a2c9"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-15 02:14:30.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-15 02:16:42.000000000","tz":600},"subject":"Remove linaro-london cloud","message":"Remove linaro-london cloud\n\nThis has been set to 0 for some time, and after discussion with kevinz\nit has been determined this cloud won\u0027t come back.  Remove it.\n\nChange-Id: Ic19ffc6dda130ecba2e70bed6bac9fe7a8b8cba9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6b332b0d2300e558c8bc1bbf928eaa552489b38f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6b332b0d2300e558c8bc1bbf928eaa552489b38f"}]},"branch":"refs/heads/master"},"1bb58d216d7b8560af5abf49994abb3be304a27e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-15 03:21:16.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/32/728332/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/32/728332/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/728332/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/728332/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/728332/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/32/728332/2"}}},"commit":{"parents":[{"commit":"c972ce4ee4fc061d6362fa3d15e28d820ebd8790","subject":"Remove Citycloud from grafana","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c972ce4ee4fc061d6362fa3d15e28d820ebd8790"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-15 02:14:30.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-15 03:17:51.000000000","tz":600},"subject":"Remove linaro-london cloud","message":"Remove linaro-london cloud\n\nThis has been set to 0 for some time, and after discussion with kevinz\nit has been determined this cloud won\u0027t come back.  Remove it.\n\nChange-Id: Ic19ffc6dda130ecba2e70bed6bac9fe7a8b8cba9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1bb58d216d7b8560af5abf49994abb3be304a27e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1bb58d216d7b8560af5abf49994abb3be304a27e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
