)]}'
{"id":"openstack%2Fproject-config~728902","triplet_id":"openstack%2Fproject-config~master~I8f4b04924e42d93e4bde39b23651128d762d7331","project":"openstack/project-config","branch":"master","topic":"retire-tricircle","hashtags":[],"change_id":"I8f4b04924e42d93e4bde39b23651128d762d7331","subject":"Retire Tricircle projects: finish infra todo","status":"MERGED","created":"2020-05-18 14:06:07.000000000","updated":"2020-06-16 08:07:48.000000000","submitted":"2020-06-16 06:42:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728902-1592289764674-25594a5d","meta_rev_id":"acdc3d3a54eb8493141f42c833c5ee0179e8e15b","_number":728902,"virtual_id_number":728902,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-16 06:42:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-06-16 06:25:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-15 22:31:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-06-16 06:25:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-27 11:33:04.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-05-30 03:00:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-15 22:31:18.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-06-16 06:25:36.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-06-16 06:42:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9244f5de279f578adf08900caac2b7ad435a8c29","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-18 14:06:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5168d12926a53945bc80c62aa57fd6c3c2d38264","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-18 14:06:16.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"1e40b2e70d063d55da6e035a1c799d28dfa5573a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 14:20:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5d5bd908e99a4a5595f7639aed9b1455 : SUCCESS in 10m 33s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7080483aa3af4deea6a8261d1dcd90bd : SUCCESS in 6m 48s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/3a1c9aec027f45188e8451b07edad7fc : SUCCESS in 11m 47s","accounts_in_message":[],"_revision_number":1},{"id":"2387edb1053c998d8e27105f7edf198c8e81ce16","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-18 14:45:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5bd5929420c0c93640608a06ee89844e8b5440d3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-18 14:45:24.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"9508d95d30e4694c985f64ccba39659f48e0ca4b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-18 14:46:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6b8a66727b8eb15c64fd1c939614132afbe72fd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 15:04:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/574a6d40c8ef4d8e84408c8629340e4e : SUCCESS in 11m 30s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a694697c697e488cb5a86792a6ca9c4b : SUCCESS in 5m 16s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/af26ff06694b4a7bb4fc35c564ca7106 : SUCCESS in 9m 27s","accounts_in_message":[],"_revision_number":3},{"id":"7ebd111054b1d60859ae5956093a5182643a0823","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-27 11:33:04.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"dcdf99c76b64b73f781eff93245238af8347cc85","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-15 18:27:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"469a4a0abe6702fc557086d6970955462148f7cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 18:53:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/03a963ee4d164a8eaa8109e6d953be16 : SUCCESS in 10m 29s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d78db1a6ac734241ae9521c278f25eb7 : SUCCESS in 9m 58s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/62ffd43ec4f5498d83286c3165b311c3 : SUCCESS in 10m 13s","accounts_in_message":[],"_revision_number":4},{"id":"56c87c2c54728ea27c974eec2be2b0348273afeb","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-06-15 22:31:18.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c1890f1a6f1fb7320ab8ecf3f7850db4d777627b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-16 06:25:36.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"80e540ec0ea365ffb13a5ab693ee3b699d2c3a7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 06:25:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e5f87a8cf07f1eaef50f1c4218e8c602845d4e9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 06:42:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"a1a07e36bdc277558562267c88ec5bf1083fb735","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 06:42:44.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e67726e0b6a440b3bba5cf4f90c5f8b1 : SUCCESS in 10m 46s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/fa0cfa004bb349529933f8ae9ed07fd3 : SUCCESS in 6m 53s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/78a3ccec88c343b2aae7efd83066d9e9 : SUCCESS in 9m 49s","accounts_in_message":[],"_revision_number":4},{"id":"24d6eb19ba7b11054ef578d8c87e350cfb13bdeb","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 08:07:48.000000000","message":"Patch Set 4:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/8b7725f878b148b59c1613b7366cbe2b : TIMED_OUT in 31m 22s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"515cd465fd756c8d1f074391f3c05924cd60b63f","revisions":{"dded86f2da968039d4c12442099958d58e266087":{"kind":"REWORK","_number":1,"created":"2020-05-18 14:06:07.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/02/728902/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/02/728902/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/728902/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/728902/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/728902/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/02/728902/1"}}},"commit":{"parents":[{"commit":"571297565313141814420097cb831800ba030219","subject":"Merge \"site-variables: remove opendev.org mirror switch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/571297565313141814420097cb831800ba030219"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-18 14:02:19.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-18 14:06:06.000000000","tz":-300},"subject":"Retire Tricircle projects: finish infra todo","message":"Retire Tricircle projects: finish infra todo\n\nTricirle project is retired now. This commit finish\nthe infra cleanup for the deliverables under Tricircle\nproject.\n\nDepends-On: https://review.opendev.org/#/c/728886/\nDepends-On: https://review.opendev.org/#/c/728890/\n\nChange-Id: I8f4b04924e42d93e4bde39b23651128d762d7331\nNeeded-By: https://review.opendev.org/#/c/728890/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dded86f2da968039d4c12442099958d58e266087"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dded86f2da968039d4c12442099958d58e266087"}]},"branch":"refs/heads/master"},"9b6dfdd91bfca1b551268dd3ec27e55c6c4bcd03":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-18 14:45:21.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/02/728902/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/02/728902/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/728902/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/728902/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/728902/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/02/728902/2"}}},"commit":{"parents":[{"commit":"76852b73d82f485efa5498849414f62dfa2285a0","subject":"Retire Tricircle project: end project gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/76852b73d82f485efa5498849414f62dfa2285a0"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-18 14:02:19.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-18 14:45:21.000000000","tz":0},"subject":"Retire Tricircle projects: finish infra todo","message":"Retire Tricircle projects: finish infra todo\n\nTricirle project is retired now. This commit finish\nthe infra cleanup for the deliverables under Tricircle\nproject.\n\nDepends-On: https://review.opendev.org/#/c/728886/\nDepends-On: https://review.opendev.org/#/c/728890/\n\nChange-Id: I8f4b04924e42d93e4bde39b23651128d762d7331\nNeeded-By: https://review.opendev.org/#/c/728890/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9b6dfdd91bfca1b551268dd3ec27e55c6c4bcd03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9b6dfdd91bfca1b551268dd3ec27e55c6c4bcd03"}]},"branch":"refs/heads/master"},"1ec4a0c9a3d659cd7da8fa2dbb634f093835fed4":{"kind":"REWORK","_number":3,"created":"2020-05-18 14:46:23.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/02/728902/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/02/728902/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/728902/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/728902/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/728902/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/02/728902/3"}}},"commit":{"parents":[{"commit":"76852b73d82f485efa5498849414f62dfa2285a0","subject":"Retire Tricircle project: end project gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/76852b73d82f485efa5498849414f62dfa2285a0"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-18 14:02:19.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-18 14:46:00.000000000","tz":-300},"subject":"Retire Tricircle projects: finish infra todo","message":"Retire Tricircle projects: finish infra todo\n\nTricirle project is retired now. This commit finish\nthe infra cleanup for the deliverables under Tricircle\nproject.\n\nDepends-On: https://review.opendev.org/#/c/728886/\nDepends-On: https://review.opendev.org/#/c/728890/\n\nChange-Id: I8f4b04924e42d93e4bde39b23651128d762d7331\nNeeded-By: https://review.opendev.org/#/c/728890/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1ec4a0c9a3d659cd7da8fa2dbb634f093835fed4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1ec4a0c9a3d659cd7da8fa2dbb634f093835fed4"}]},"branch":"refs/heads/master"},"515cd465fd756c8d1f074391f3c05924cd60b63f":{"kind":"REWORK","_number":4,"created":"2020-06-15 18:27:08.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/02/728902/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/02/728902/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/728902/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/728902/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/728902/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/02/728902/4"}}},"commit":{"parents":[{"commit":"a401f4e5255588d07d17449b8294990c7e34875d","subject":"Merge \"Create a new project for recordsansible/ara-collection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a401f4e5255588d07d17449b8294990c7e34875d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-15 18:26:33.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-15 18:26:33.000000000","tz":-300},"subject":"Retire Tricircle projects: finish infra todo","message":"Retire Tricircle projects: finish infra todo\n\nTricirle project is retired now. This commit finish\nthe infra cleanup for the deliverables under Tricircle\nproject.\n\nDepends-On: https://review.opendev.org/#/c/728886/\nDepends-On: https://review.opendev.org/#/c/728890/\nNeeded-By: https://review.opendev.org/#/c/728890/\n\nChange-Id: I8f4b04924e42d93e4bde39b23651128d762d7331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/515cd465fd756c8d1f074391f3c05924cd60b63f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/515cd465fd756c8d1f074391f3c05924cd60b63f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
