)]}'
{"id":"openstack%2Fproject-config~728933","triplet_id":"openstack%2Fproject-config~master~I2613f3bc122c187ae944e0ab23246e92be4c3faa","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I2613f3bc122c187ae944e0ab23246e92be4c3faa","subject":"Add Ubuntu Focal for AArch64","status":"MERGED","created":"2020-05-18 16:09:39.000000000","updated":"2020-05-18 23:13:28.000000000","submitted":"2020-05-18 23:05:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728933-1589843120383-8f3df06f","meta_rev_id":"bcbb47a6ba8001784655dcaf6764b04f5eb2a448","_number":728933,"virtual_id_number":728933,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-05-18 18:41:19.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-18 23:05:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-05-18 18:47:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-18 22:58:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"date":"2020-05-18 19:52:06.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-18 22:58:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-18 18:47:50.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-05-18 19:52:06.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-18 22:58:10.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-05-18 23:05:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f697f475709107515092b2fece207bb1314dd116","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-18 16:09:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec52dde558fabd5e76cd739ecb77c6c7ab8ca917","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 16:21:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e7e74ff47f9b491797a26c27dc04a921 : SUCCESS in 8m 31s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/6b0ff1a909c34eba9ae95bace50ca992 : SUCCESS in 4m 11s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/ce0ce1d79c8142a6aa8ba064bd637a5c : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":1},{"id":"585be81b06fa22237ad312d6b373171100c99d40","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-18 18:35:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4b45f41c587048ae77babb81340ade6816e2766a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-18 18:41:19.000000000","message":"Patch Set 2:\n\nFocal has librdkafka 1.4.x and newer Cython3 so we can use it to build all wheels used by openstack/requirements.\n\nhttps://review.opendev.org/728798 job will be switched to Focal once it  be available.","accounts_in_message":[],"_revision_number":2},{"id":"22d7002e17e5a850828e1f5d94aacf8a3bb89628","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-05-18 18:47:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"925b0d4a15274247f76ec9c6ac23112932e5e706","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 18:48:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8552365151a0460a9e5e2e96f2b39ada : SUCCESS in 8m 09s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/16ace68e58e54e659c805c658c996c8b : SUCCESS in 4m 51s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/6620183b436e4bb4874fbd62b254d5e2 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":2},{"id":"5c66dc56b05bc06d951e84b942a6a2f3b50a7102","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-18 22:58:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nI have https://review.opendev.org/#/c/725752/ in progress to cover this in dib, but it is passing, so we should be good to build.","accounts_in_message":[],"_revision_number":2},{"id":"f022878ab9ff979eff00191241ff10c1c353a421","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 22:58:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"10aca010eff86adf7dc0c13d359caa01f1350bbd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 23:05:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"bea3f05840c538ca57302a338dcd9d72a18df63a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 23:05:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fc2270a1c10c4ad1b42c38a5cc86126e : SUCCESS in 6m 46s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/a8ba9f66c9be4887ac061a02248bd594 : SUCCESS in 4m 23s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/69cf20abe78e42b39ee4a4d5c6bfe94e : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":2},{"id":"bcbb47a6ba8001784655dcaf6764b04f5eb2a448","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 23:13:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/bfe2ff9a657245afa244b6677e212979 : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"18c490d15a92978d915000ebe35151d1218fbe1c","revisions":{"d90fd71d66404627fdd151e8fbea7a32429e6794":{"kind":"REWORK","_number":1,"created":"2020-05-18 16:09:39.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/33/728933/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/33/728933/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/728933/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/728933/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/728933/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/33/728933/1"}}},"commit":{"parents":[{"commit":"571297565313141814420097cb831800ba030219","subject":"Merge \"site-variables: remove opendev.org mirror switch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/571297565313141814420097cb831800ba030219"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-18 16:08:57.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-18 16:08:57.000000000","tz":120},"subject":"Add Ubuntu Focal for AArch64","message":"Add Ubuntu Focal for AArch64\n\nubuntu-focal-arm64 in all places where ubuntu-bionic-arm64 is present\n\nChange-Id: I2613f3bc122c187ae944e0ab23246e92be4c3faa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d90fd71d66404627fdd151e8fbea7a32429e6794"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d90fd71d66404627fdd151e8fbea7a32429e6794"}]},"branch":"refs/heads/master"},"18c490d15a92978d915000ebe35151d1218fbe1c":{"kind":"REWORK","_number":2,"created":"2020-05-18 18:35:29.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/33/728933/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/33/728933/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/728933/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/728933/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/728933/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/33/728933/2"}}},"commit":{"parents":[{"commit":"60f352bfced1a99a0b098313aa113adf4a08ae1d","subject":"Use infra-root-keys-2020-05-13 in nodepool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/60f352bfced1a99a0b098313aa113adf4a08ae1d"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-18 16:08:57.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-18 18:35:06.000000000","tz":120},"subject":"Add Ubuntu Focal for AArch64","message":"Add Ubuntu Focal for AArch64\n\nubuntu-focal-arm64 in all places where ubuntu-bionic-arm64 is present\n\nChange-Id: I2613f3bc122c187ae944e0ab23246e92be4c3faa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/18c490d15a92978d915000ebe35151d1218fbe1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/18c490d15a92978d915000ebe35151d1218fbe1c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
