)]}'
{"id":"openstack%2Fproject-config~729334","triplet_id":"openstack%2Fproject-config~master~Id0d49459d84ad9540ef373a838a01ca0ca06486c","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Id0d49459d84ad9540ef373a838a01ca0ca06486c","subject":"Configure ansible-role-lunasa-hsm for release","status":"MERGED","created":"2020-05-19 17:35:26.000000000","updated":"2020-05-25 08:02:11.000000000","submitted":"2020-05-25 06:25:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729334-1590387934411-7d9b3224","meta_rev_id":"2d1f2aaad00f96acd3dad34cf5d07dacc70a8d6a","_number":729334,"virtual_id_number":729334,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-25 06:25:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-05-20 04:50:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-25 06:04:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-25 06:04:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-20 04:50:41.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-25 06:04:15.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-05-25 06:25:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da44905ba1ec08da788b840503acdf7350aa30e5","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-05-19 17:35:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f05aea50495a8434e102d2b6297990d2af09379","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 17:47:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/39e17fe4201d4bcd89a5ee684c67f68e : FAILURE in 5m 16s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/953482aa0c374318bfa93db41533fc88 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":1},{"id":"0352bb0439d530c363ac8fb3043877019ae64361","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-05-19 19:12:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"98e77d44d35c883e08c0de118284a395571876d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 19:25:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c337b6540cd9448db30448ea59754afc : FAILURE in 7m 11s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c059396f6a534711bf8a3c9af1189515 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":2},{"id":"74ce4abb9ee050585fcb3382ced7a6c6814eaeb2","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-05-19 19:31:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c727fa3f07f55e3b94d8b7fddbca2bf70ea87b02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 19:43:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b6d312665fda4b4cb3d1c72222edd861 : FAILURE in 5m 12s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d8cb8a6bd99e4c13b460977f9fd38ff8 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":3},{"id":"b2e40e985988429d1d64445453208cfc8e6183b6","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2020-05-20 00:05:41.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"d06a1392a52ec21d5930be9d80c270ea9c0c3e79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 00:14:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/26258d9ee0494b239d8678cf0742ec68 : SUCCESS in 6m 41s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/526be8d483b24027b18bf0ae9b15987b : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":4},{"id":"a214594050690223fa53081d1d8424a3d87430c8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-20 04:50:41.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"532272b1fc79f57c61f9df06dff82917ab8c6da8","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-25 06:04:15.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"0af950266296c593136b9bd9a82274eed8bb5889","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 06:04:33.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"093094b7512a864a187c233bb86980c60386d820","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 06:25:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"5e6b0dae6b72f8c1f6b3adefb73d2d4fa75e7216","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 06:25:34.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/382da6ed305c47848f2cc7c666b6101e : SUCCESS in 8m 39s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c91abf69b78b412cbf96b57211e0e5d7 : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":4},{"id":"2d1f2aaad00f96acd3dad34cf5d07dacc70a8d6a","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 08:02:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/0562a528c82d46cba4f7823f13ab940e : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"965410963c0e576c5ee501dc83553d1e6f51a43e","revisions":{"e52f2ec6d8344ae7e41a39f0beaf5f8601f81a95":{"kind":"REWORK","_number":1,"created":"2020-05-19 17:35:26.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/34/729334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/34/729334/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/729334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/729334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/729334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/34/729334/1"}}},"commit":{"parents":[{"commit":"1bd14f4eeaeb798fa6c7347e3f1bed6fbcf842a4","subject":"Merge \"grafana: add CentOS 8 and missing AArch64 entries\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1bd14f4eeaeb798fa6c7347e3f1bed6fbcf842a4"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-05-19 17:35:08.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-05-19 17:35:08.000000000","tz":-300},"subject":"Add storyboard project for ansible-role-lunasa-hsm","message":"Add storyboard project for ansible-role-lunasa-hsm\n\nChange-Id: Id0d49459d84ad9540ef373a838a01ca0ca06486c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e52f2ec6d8344ae7e41a39f0beaf5f8601f81a95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e52f2ec6d8344ae7e41a39f0beaf5f8601f81a95"}]},"branch":"refs/heads/master"},"977e5437827c73032f2912263fa8f48be65c09e5":{"kind":"REWORK","_number":2,"created":"2020-05-19 19:12:00.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/34/729334/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/34/729334/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/729334/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/729334/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/729334/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/34/729334/2"}}},"commit":{"parents":[{"commit":"1bd14f4eeaeb798fa6c7347e3f1bed6fbcf842a4","subject":"Merge \"grafana: add CentOS 8 and missing AArch64 entries\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1bd14f4eeaeb798fa6c7347e3f1bed6fbcf842a4"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-05-19 17:35:08.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-05-19 19:11:52.000000000","tz":-300},"subject":"Add storyboard project for ansible-role-lunasa-hsm","message":"Add storyboard project for ansible-role-lunasa-hsm\n\nChange-Id: Id0d49459d84ad9540ef373a838a01ca0ca06486c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/977e5437827c73032f2912263fa8f48be65c09e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/977e5437827c73032f2912263fa8f48be65c09e5"}]},"branch":"refs/heads/master"},"1a461ea49799d8db9169edb2290c7044753a69d4":{"kind":"REWORK","_number":3,"created":"2020-05-19 19:31:48.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/34/729334/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/34/729334/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/729334/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/729334/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/729334/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/34/729334/3"}}},"commit":{"parents":[{"commit":"1bd14f4eeaeb798fa6c7347e3f1bed6fbcf842a4","subject":"Merge \"grafana: add CentOS 8 and missing AArch64 entries\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1bd14f4eeaeb798fa6c7347e3f1bed6fbcf842a4"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-05-19 17:35:08.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-05-19 19:29:12.000000000","tz":-300},"subject":"Configure ansible-role-lunasa-hsm for release","message":"Configure ansible-role-lunasa-hsm for release\n\nThis patch adds the publish-to-pypi template to the ansible\nrole repo, as well as a storyboard project to track issues.\n\nChange-Id: Id0d49459d84ad9540ef373a838a01ca0ca06486c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1a461ea49799d8db9169edb2290c7044753a69d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1a461ea49799d8db9169edb2290c7044753a69d4"}]},"branch":"refs/heads/master"},"965410963c0e576c5ee501dc83553d1e6f51a43e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-05-20 00:05:41.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/34/729334/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/34/729334/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/729334/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/729334/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/729334/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/34/729334/4"}}},"commit":{"parents":[{"commit":"e25845d8faaff52cdc184df2b1372f519ce75b88","subject":"Escape use of % in tox.ini to avoid interpolation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e25845d8faaff52cdc184df2b1372f519ce75b88"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-05-19 17:35:08.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2020-05-20 00:05:41.000000000","tz":0},"subject":"Configure ansible-role-lunasa-hsm for release","message":"Configure ansible-role-lunasa-hsm for release\n\nThis patch adds the publish-to-pypi template to the ansible\nrole repo, as well as a storyboard project to track issues.\n\nChange-Id: Id0d49459d84ad9540ef373a838a01ca0ca06486c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/965410963c0e576c5ee501dc83553d1e6f51a43e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/965410963c0e576c5ee501dc83553d1e6f51a43e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
