)]}'
{"id":"openstack%2Fproject-config~729412","triplet_id":"openstack%2Fproject-config~master~I7e23595495909fc94a7ea5a9e461e67e25ddeae9","project":"openstack/project-config","branch":"master","topic":"escape-percent-tox.ini","hashtags":[],"change_id":"I7e23595495909fc94a7ea5a9e461e67e25ddeae9","subject":"Escape use of % in tox.ini to avoid interpolation","status":"MERGED","created":"2020-05-19 22:16:15.000000000","updated":"2020-05-19 23:12:51.000000000","submitted":"2020-05-19 23:09:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729412-1589929771862-472850b6","meta_rev_id":"56c087c4c7e346208b616f21b97df02a1bb96573","_number":729412,"virtual_id_number":729412,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-19 23:09:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2020-05-19 22:39:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-19 22:52:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-19 22:52:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-19 22:39:29.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2020-05-19 22:52:41.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-05-19 23:09:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9f87623e3cc5de354de824205826cc16468c4de7","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-05-19 22:16:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac1387b0fe31744d936c0e8a04d2854d9a3e6edd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 22:31:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/9a8ff44cf12d4ca18a52b1647f2a2b66 : SUCCESS in 4m 02s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/72fb42c3777a4229aced87da3931fd99 : SUCCESS in 8m 48s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/f618750d189c4a368ccda9187fa226f7 : SUCCESS in 4m 09s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/24d4387b46fe4d89a61ecebe6f5d8cca : SUCCESS in 5m 51s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/43bad767f1c04d7294fe4a829732d691 : SUCCESS in 5m 01s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/e589bd069fba433b8e1ea2b72ffcb112 : SUCCESS in 4m 01s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/6a6ca4b053044b40bfa7ccaa6cf92f30 : SUCCESS in 9m 16s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/bd0764dc4bb14f96b6f823902cc30dd3 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":1},{"id":"18554420191d115cea71182e1283ed8406852950","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-05-19 22:39:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"06309e133e1764d3d89bb9f2a269259b0a895e9c","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-19 22:52:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7bb088e00eec6e2f8f1c82572bbdfb03841c8695","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 22:52:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9e24153f58ab3f3677df107ffdbc6df1873c32ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 23:09:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e8ad0ed2ace09136ad10aec5c226b4f70f2d4970","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 23:09:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/9f2013b0d7274af18dc084c39e2506f1 : SUCCESS in 4m 22s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a4b5c3cd82f94d88b3a38f6fb90f91a1 : SUCCESS in 9m 49s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/4bb5bbeed969484389c3cad5e742f9cf : SUCCESS in 5m 46s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/15e4fadfb886442f8f153b44b4c8298a : SUCCESS in 6m 13s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/d658d80424e048ae82f2c3271d7a0ba1 : SUCCESS in 5m 28s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/f950d3e791a14b62b0dd4f1952a56703 : SUCCESS in 5m 54s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/71978795f7a84794825177e2663b4d8d : SUCCESS in 10m 49s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/4e568a3183e145fa88a8b4dc8a5248b4 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"c7513ab67ad4609c0324a0f8c3d2b7e9e0b12095","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 23:12:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- promote-infra-index https://zuul.opendev.org/t/openstack/build/3897636e9d0d43b5a9383f958bd51f5c : SUCCESS in 1m 54s\n- promote-openstack-specs-site https://zuul.opendev.org/t/openstack/build/25d6f0137a3f4a3fa0095ffdec3b0de5 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e25845d8faaff52cdc184df2b1372f519ce75b88","revisions":{"e25845d8faaff52cdc184df2b1372f519ce75b88":{"kind":"REWORK","_number":1,"created":"2020-05-19 22:16:15.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/12/729412/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/12/729412/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/729412/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/729412/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/729412/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/12/729412/1"}}},"commit":{"parents":[{"commit":"1bd14f4eeaeb798fa6c7347e3f1bed6fbcf842a4","subject":"Merge \"grafana: add CentOS 8 and missing AArch64 entries\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1bd14f4eeaeb798fa6c7347e3f1bed6fbcf842a4"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-05-19 22:15:05.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-05-19 22:15:05.000000000","tz":-420},"subject":"Escape use of % in tox.ini to avoid interpolation","message":"Escape use of % in tox.ini to avoid interpolation\n\nZuul-jobs tox roles wants to be able to run tox --showconfig -e linters.\nUnfortunately our linters target has an unescaped % symbol in it which\ncauses python3 + tox to fail with an interpolation error. Escaping the %\nto avoid interpolation addresses this.\n\nChange-Id: I7e23595495909fc94a7ea5a9e461e67e25ddeae9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e25845d8faaff52cdc184df2b1372f519ce75b88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e25845d8faaff52cdc184df2b1372f519ce75b88"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
